)]}'
{"id":"openstack%2Fproject-config~840594","triplet_id":"openstack%2Fproject-config~master~Ie00c42857e7a541aa1ccb8fe480756c1e7a38dc4","project":"openstack/project-config","branch":"master","topic":"opendev-openafs-deb","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-05-05 22:37:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie00c42857e7a541aa1ccb8fe480756c1e7a38dc4","subject":"Add opendev/infra-openafs-deb","status":"MERGED","created":"2022-05-05 09:13:02.000000000","updated":"2022-05-05 22:58:47.000000000","submitted":"2022-05-05 22:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840594-opendev-openafs-deb","meta_rev_id":"975309f87f691001e70c706b65dc486b2df81bee","_number":840594,"virtual_id_number":840594,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-05 22:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-05-05 22:28:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-05 12:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2022-05-05 22:28:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 09:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 12:31:18.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-05-05 22:28:35.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"02146d6b4063d52faf94f17d1a9a1e1f95ee0b8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-05-05 09:13:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d36d6d0bf28a1e83d2f6879371feb38663c63c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 09:25:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/51f54b38e7f44976b012c3e8b098b2c0 : SUCCESS in 7m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b3229000b4c84b70bae5af8b41927e58 : SUCCESS in 3m 34s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c6aebc2160ed41949a4d5604b6fd1661 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b3c7f44fc3f09d3bc799f5b0dfb8ba1d59761ce9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-05-05 09:25:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8a009cef62390b06a04d33f500763ff646af6b60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 09:34:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c081558da7bd4f0fae8e524eb3c0f109 : SUCCESS in 7m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/11ad1d45ee2f476fac5d0f62f591284c : SUCCESS in 4m 00s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/cd5e2a8c0aae40b7a69a8d7512813674 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":2},{"id":"de4ee6e355226339b1180e9760802a4a7eaddfbe","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-05-05 12:31:18.000000000","message":"Patch Set 2: Code-Review+2\n\nThis gets us closer to the workflow described at https://salsa.debian.org/debian/openafs/-/blob/master/debian/README.source too (note that salsa itself contains per-distro-release branches of the debian file tree for the openafs source package).","accounts_in_message":[],"_revision_number":2},{"id":"c02661ee4c9ffa51084641505fdecaecf1a46a39","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-05-05 22:28:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d845856dc6f6fc9ad67f98921212f722b3660e46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 22:29:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"01c651662fbffe595d36ea3c77a1c484a39dde5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 22:37:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d69a313a18024941a178f53fb0e2b922 : SUCCESS in 7m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f3034f592a5a4a1783a0c1c2353754b7 : SUCCESS in 4m 11s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c4760c91631b46a2984f39f2cf14d878 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7a32f9ef1a957c889d1bb5bcf81168f4ae6df8ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 22:37:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"975309f87f691001e70c706b65dc486b2df81bee","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 22:58:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/be254eb931034b7189636daf185f1bf0 : SUCCESS in 7m 19s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/c3671b86d7534940a50473a2887e54ff : SUCCESS in 7m 56s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/210efead195f4c5e94065bb1e74d1a65 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec06a200acea649db4a12bf194dc352a23a1ed47","revisions":{"0bb55edf17ba73ed0ac35bd2d55e61d46d591ff7":{"kind":"REWORK","_number":1,"created":"2022-05-05 09:13:02.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/840594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/840594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/840594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/840594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/840594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/840594/1"}}},"commit":{"parents":[{"commit":"2f451b3dac724480a6d51a29c9317f06fdc645dd","subject":"Merge \"Add Jammy arm64 images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f451b3dac724480a6d51a29c9317f06fdc645dd"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-05 09:04:56.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-05 09:04:56.000000000","tz":600},"subject":"Add opendev/infra-openafs-deb","message":"Add opendev/infra-openafs-deb\n\nThis is a project to hold things related to building the OpenAFS\npackages used by OpenDev infra.  These packages are pushed to a PPA\nwhich is consumed by our production machines.\n\nWe keep a separate PPA for these packages as we have had occasions in\nthe past where we either needed more recent versions of OpenAFS for\nvarious platform support, or had to apply patches with some urgency to\nkeep things going.  Being able to quickly produce new packages has\nbeen very helpful.\n\nPreviously these have been created by hand, where somebody uploads the\ndebian package bits to the PPA manually and a .deb is produced.  Of\ncourse the source of those packages is available, but it\u0027s not very\ngit-ops-y or collaborative.\n\nThis started in I07780595331da29af9f72185cbb3499210e89c19 where\nfungi\u0027s concerns that I argued against were completely correct.\nTrying to build across xenial -\u003e jammy with one debian/* directory is\nnot going to work, and checking in multiple trees is very unwieldy in\nthat repo.\n\nThis proposes a separate repo where we can keep branches for each\ndistro\u0027s build.  We can then use the existing infrastructure we\u0027ve\nbuilt to build the source packages and push to the PPA for building.\n\nChange-Id: Ie00c42857e7a541aa1ccb8fe480756c1e7a38dc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0bb55edf17ba73ed0ac35bd2d55e61d46d591ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0bb55edf17ba73ed0ac35bd2d55e61d46d591ff7"}]},"branch":"refs/heads/master"},"ec06a200acea649db4a12bf194dc352a23a1ed47":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-05 09:25:28.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/840594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/840594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/840594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/840594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/840594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/840594/2"}}},"commit":{"parents":[{"commit":"2f451b3dac724480a6d51a29c9317f06fdc645dd","subject":"Merge \"Add Jammy arm64 images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f451b3dac724480a6d51a29c9317f06fdc645dd"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-05 09:04:56.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-05 09:22:18.000000000","tz":600},"subject":"Add opendev/infra-openafs-deb","message":"Add opendev/infra-openafs-deb\n\nThis is a project to hold things related to building the OpenAFS\npackages used by OpenDev infra.  These packages are pushed to a PPA\nwhich is consumed by our production machines.\n\nWe keep a separate PPA for these packages as we have had occasions in\nthe past where we either needed more recent versions of OpenAFS for\nvarious platform support, or had to apply patches with some urgency to\nkeep things going.  Being able to quickly produce new packages has\nbeen very helpful.\n\nPreviously these have been created by hand, where somebody uploads the\ndebian package bits to the PPA manually and a .deb is produced.  Of\ncourse the source of those packages is available, but it\u0027s not very\ngit-ops-y or collaborative.\n\nThis started in I07780595331da29af9f72185cbb3499210e89c19 where\nfungi\u0027s concerns that I argued against were completely correct.\nTrying to build across xenial -\u003e jammy with one debian/* directory is\nnot going to work, and checking in multiple trees is very unwieldy in\nthat repo.\n\nThis proposes a separate repo where we can keep branches for each\ndistro\u0027s build.  We can then use the existing infrastructure we\u0027ve\nbuilt to build the source packages and push to the PPA for building.\n\nI\u0027ve chosen the name with \"infra-\" in front to make it clear these\naren\u0027t supposed to be general purpose packages in any way, but are for\nopendev infra use.  I\u0027ve put this in the openstack tenant because the\nbuild bits are still in openstack/openstack-zuul-jobs for historical\nreasons -- this is where we do similar builds for RPM (which uses AFS\npublishing jobs, and has openstack embedded in paths, etc.) and it\nmakes sense to keep them together.  We can consider moving everything\nunder the opendev tenant in the future.\n\nChange-Id: Ie00c42857e7a541aa1ccb8fe480756c1e7a38dc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec06a200acea649db4a12bf194dc352a23a1ed47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec06a200acea649db4a12bf194dc352a23a1ed47"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
