)]}'
{"id":"openstack%2Fproject-config~841190","triplet_id":"openstack%2Fproject-config~master~Ida6e34feb2ca3d7e4bb7a09610aa69dec83b8dc0","project":"openstack/project-config","branch":"master","topic":"vhd-utils-ppa","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-05-10 04:52:18.000000000","reason":"Ian Wienand replied on the change"}},"hashtags":[],"change_id":"Ida6e34feb2ca3d7e4bb7a09610aa69dec83b8dc0","subject":"infra-vhd-util-deb : add jobs","status":"MERGED","created":"2022-05-10 02:01:34.000000000","updated":"2022-05-10 04:59:22.000000000","submitted":"2022-05-10 04:59:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841190-vhd-utils-ppa","meta_rev_id":"92fe4a08b943487e59e71df5294111d6a636f474","_number":841190,"virtual_id_number":841190,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 04:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-10 04:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-10 04:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-10 02:06:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7143c9bf8b259213b870f18ba18f77deb35aee8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-05-10 02:01:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7053a4506ced2cd30a57bba5a31f5a30966d73c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 02:06:44.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job vhd-util-package-build not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/infra-vhd-util-deb\n      check:\n        jobs:\n          - noop:\n              branches: master\n          - vhd-util-package-build:\n              branches: ^(?!master).*$\n      gate:\n        jobs:\n          - noop:\n              branches: master\n          - vhd-util-package-build:\n              branches: ^(?!master).*$\n      promote:\n        jobs:\n          - vhd-util-package-build-promote:\n              branches: ^(?!master).*$\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 180, column 3","accounts_in_message":[],"_revision_number":1},{"id":"fc38c49531980762efdafa3341d32366780ceb91","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-05-10 03:44:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df109fa54e80ed76760a00810e054fd7ed729ae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 03:44:44.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job vhd-util-package-build not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/infra-vhd-util-deb\n      check:\n        jobs:\n          - noop:\n              branches: master\n          - vhd-util-package-build:\n              branches: ^(?!master).*$\n      gate:\n        jobs:\n          - noop:\n              branches: master\n          - vhd-util-package-build:\n              branches: ^(?!master).*$\n      promote:\n        jobs:\n          - vhd-util-package-build-promote:\n              branches: ^(?!master).*$\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 180, column 3","accounts_in_message":[],"_revision_number":1},{"id":"e98a5e3d885dc60635d905da2b2d559e8eabfc6a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-05-10 04:39:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7ad30fbc42c250bdb5cbfa790285d3d89150a0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 04:40:08.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Job vhd-util-package-build not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/infra-vhd-util-deb\n      check:\n        jobs:\n          - noop:\n              branches: master\n          - vhd-util-package-build:\n              branches: ^(?!master).*$\n      gate:\n        jobs:\n          - noop:\n              branches: master\n          - vhd-util-package-build:\n              branches: ^(?!master).*$\n      promote:\n        jobs:\n          - vhd-util-package-build-promote:\n              branches: ^(?!master).*$\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 180, column 3","accounts_in_message":[],"_revision_number":1},{"id":"1d87ca566e407f7726134ee10d223889b9da78ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-05-10 04:42:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"326c9e986759053a7266a7cccfa1580b1db1d093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 04:51:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/01bfd5a2f79e4338a4e254fedeec8e35 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":2},{"id":"0f4aa9be7befd1f4c28e013d66e2eb63b7b152c0","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-05-10 04:52:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3058ad6c3a675c64a920fd7e5f3fe092334bc70c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 04:52:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5550977b3ecca5016335210211df65242d40fc6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 04:59:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c43fc5ec1eb84073bf2599f7078d10c4 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":2},{"id":"92fe4a08b943487e59e71df5294111d6a636f474","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 04:59:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"739c5a6002be0dceaf9b9aa1ce0d52547130deda","revisions":{"22c02c74ce697bfea22ca662327104bec8372014":{"kind":"REWORK","_number":1,"created":"2022-05-10 02:01:34.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/90/841190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/841190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/841190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/841190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/841190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/841190/1"}}},"commit":{"parents":[{"commit":"00080f07923173136568c07ad19dcee2546ccf6f","subject":"Add opendev/infra-vhd-util-deb repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00080f07923173136568c07ad19dcee2546ccf6f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-10 02:00:47.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-10 02:00:47.000000000","tz":600},"subject":"infra-vhd-util-deb : add jobs","message":"infra-vhd-util-deb : add jobs\n\nAdd the jobs for infra-vhd-util-deb (the publish job needs to be here,\nwe keep the rest just to keep it together).\n\nChange-Id: Ida6e34feb2ca3d7e4bb7a09610aa69dec83b8dc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22c02c74ce697bfea22ca662327104bec8372014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/22c02c74ce697bfea22ca662327104bec8372014"}]},"branch":"refs/heads/master"},"739c5a6002be0dceaf9b9aa1ce0d52547130deda":{"kind":"REWORK","_number":2,"created":"2022-05-10 04:42:32.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/90/841190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/841190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/841190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/841190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/841190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/841190/2"}}},"commit":{"parents":[{"commit":"00080f07923173136568c07ad19dcee2546ccf6f","subject":"Add opendev/infra-vhd-util-deb repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00080f07923173136568c07ad19dcee2546ccf6f"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-10 02:00:47.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-05-10 04:42:22.000000000","tz":600},"subject":"infra-vhd-util-deb : add jobs","message":"infra-vhd-util-deb : add jobs\n\nAdd the jobs for infra-vhd-util-deb (the publish job needs to be here,\nwe keep the rest just to keep it together).\n\nChange-Id: Ida6e34feb2ca3d7e4bb7a09610aa69dec83b8dc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/739c5a6002be0dceaf9b9aa1ce0d52547130deda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/739c5a6002be0dceaf9b9aa1ce0d52547130deda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
