)]}'
{"id":"openstack%2Fproject-config~844116","triplet_id":"openstack%2Fproject-config~master~I0e13d464f27084fecdaf76cf2dd750c3c6d71eab","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2023-11-08 07:37:45.000000000","reason":"Change was abandoned"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-11-08 07:37:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0e13d464f27084fecdaf76cf2dd750c3c6d71eab","subject":"Add 16 vcpu flavors","status":"ABANDONED","created":"2022-05-31 17:39:56.000000000","updated":"2023-11-08 07:37:45.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"dd2e84196318f97abbf7ab930a4006543784fa09","_number":844116,"virtual_id_number":844116,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-24 20:08:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-06 19:49:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2022-08-24 19:54:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-31 17:49:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-05 12:24:12.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-09-06 19:49:30.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2023-10-05 02:17:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"cae5f2d96bc529ca8d1225bb22d046fda9f34e65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2022-05-31 17:39:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a83b67d8d9a8b71d87fa413a67636bf9c1ec1eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 17:49:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/03ed1c888135407c9902a38e0f274b71 : SUCCESS in 7m 31s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/21ce235d47094a7fb8092ffc6f69d54a : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"eddb9bf92898160fa267e2c009aa929a698cd81e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-05 12:24:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de13421706346f11f1fc29f0e7bbac8b61c7bee1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-24 19:29:59.000000000","message":"Patch Set 1: Workflow+1\n\nI\u0027m going ahead and single-core approving this since it was proposed by another core reviewer anyway and there has been ample time for others to raise objections.","accounts_in_message":[],"_revision_number":1},{"id":"9a6ee9b31ea304593bcc53e95b57949f5569153d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 19:30:25.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/project-config for 844116,1","accounts_in_message":[],"_revision_number":1},{"id":"63c35579a67fa2da6a96e500f0ac3e0d898f8971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-24 19:54:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb070bb9971d6ff1e0303adef1a233235d2357e2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-24 19:54:51.000000000","message":"Patch Set 2: Code-Review+2\n\nRebased to address a minor merge conflict in the nl03 config.","accounts_in_message":[],"_revision_number":2},{"id":"0e90f64ce9e86757a9c21362a2f8808c4c2f73aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 20:08:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2857efb197a4ab4b46c5a18fbd7f063\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0e815acd4901418d85eee22cef0df970 : SUCCESS in 4m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/461de2e0c934475e909e2fa92887d18f : SUCCESS in 7m 25s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b3baa95c76164391b14d2212b803744e : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9b9b0cdc1139100de199f75965f7a5aed4ac8a68","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-09-06 19:49:30.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e963d1978de488112d531feacd3c9e920b3d4a2e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-09-06 19:57:12.000000000","message":"Patch Set 2:\n\nI wonder if a feature in the Nodepool/Zuul API and eventually UI to track flavor data could be useful too, so that we wouldn\u0027t need to separately maintain documentation which can get out of sync.","accounts_in_message":[],"_revision_number":2},{"id":"ca084ff07d001ef09b1f3833bfb59c57892e0a31","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-05 02:17:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd2e84196318f97abbf7ab930a4006543784fa09","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-08 07:37:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b3d3cd22e73ef41ed422dc5f1e55e7879574f434","revisions":{"01c6ef4993e46fefaf6677c1f4dd7c09a022e83d":{"kind":"REWORK","_number":1,"created":"2022-05-31 17:39:56.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/16/844116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/844116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/844116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/844116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/844116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/844116/1"}}},"commit":{"parents":[{"commit":"d391c5b04a638295eec064e0b40258789537952e","subject":"Fix kolla panel for Grafana, part 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d391c5b04a638295eec064e0b40258789537952e"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2022-05-31 17:36:00.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2022-05-31 17:39:25.000000000","tz":-420},"subject":"Add 16 vcpu flavors","message":"Add 16 vcpu flavors\n\nThis adds ubuntu-focal-16c flavors in the clouds where that is feasible:\n* rax\n* iweb\n* vexxhost\n\nLimestone, inmotion, and ovh need custom flavors added.\n\nNote that these have varying amounts of ram, so we really should\ntry to limit the ram used on boot.  Until that happens, caveat\ndeveloper.\n\nChange-Id: I0e13d464f27084fecdaf76cf2dd750c3c6d71eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01c6ef4993e46fefaf6677c1f4dd7c09a022e83d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01c6ef4993e46fefaf6677c1f4dd7c09a022e83d"}]},"branch":"refs/heads/master"},"b3d3cd22e73ef41ed422dc5f1e55e7879574f434":{"kind":"REWORK","_number":2,"created":"2022-08-24 19:54:10.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/16/844116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/16/844116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/844116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/844116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/16/844116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/16/844116/2"}}},"commit":{"parents":[{"commit":"36b9c302e5f0e5e8305ce61d455e914788a76c29","subject":"nodepool: Add Fedora 36","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36b9c302e5f0e5e8305ce61d455e914788a76c29"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2022-05-31 17:36:00.000000000","tz":-420},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-08-24 19:53:46.000000000","tz":0},"subject":"Add 16 vcpu flavors","message":"Add 16 vcpu flavors\n\nThis adds ubuntu-focal-16c flavors in the clouds where that is feasible:\n* rax\n* iweb\n* vexxhost\n\nLimestone, inmotion, and ovh need custom flavors added.\n\nNote that these have varying amounts of ram, so we really should\ntry to limit the ram used on boot.  Until that happens, caveat\ndeveloper.\n\nChange-Id: I0e13d464f27084fecdaf76cf2dd750c3c6d71eab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b3d3cd22e73ef41ed422dc5f1e55e7879574f434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b3d3cd22e73ef41ed422dc5f1e55e7879574f434"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
