)]}'
{"id":"openstack%2Fproject-config~846427","triplet_id":"openstack%2Fproject-config~master~I0456da209310e1b9d0af6f1281433b3d9edb7296","project":"openstack/project-config","branch":"master","topic":"install-virtualenv","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-06-20 22:38:05.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-06-20 17:23:40.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}}},"hashtags":[],"change_id":"I0456da209310e1b9d0af6f1281433b3d9edb7296","subject":"Install virtualenv in proposal jobs","status":"ABANDONED","created":"2022-06-17 16:30:17.000000000","updated":"2022-06-20 22:38:05.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d7e3e236d4b4f52b8a3bf7e28fe3cc3a13f3c006","_number":846427,"virtual_id_number":846427,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-17 16:41:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 16:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-20 09:06:56.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"1b3c83b72db92c25ca88c0dbc787066d835c2e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-06-17 16:30:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fa2b058e25b5dbe70f701f12cb2b995ac928db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 16:41:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e6ebf00d81c745da9fc77b38a6ae9ec4 : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"645a6b96d6f1bcf1dd54d3000db110bd781885a5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-20 09:06:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13112e3444b4bf665f910df4bd252690721373ed","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-06-20 16:43:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5843c85a551b0a957e42e9a9c7e36bedf37ae76c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-20 17:23:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7e3e236d4b4f52b8a3bf7e28fe3cc3a13f3c006","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-06-20 22:38:05.000000000","message":"Abandoned\n\nAbandoning in favor of https://review.opendev.org/c/openstack/project-config/+/846794","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4fcc8adcf61c9906bca861cff5b5f0e83a801668","revisions":{"4fcc8adcf61c9906bca861cff5b5f0e83a801668":{"kind":"REWORK","_number":1,"created":"2022-06-17 16:30:17.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/27/846427/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/846427/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/846427/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/846427/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/846427/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/846427/1"}}},"commit":{"parents":[{"commit":"e06aa44fe1ec21ff070b503e804d93c0b5726c46","subject":"Use virtualenv in translation update jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e06aa44fe1ec21ff070b503e804d93c0b5726c46"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-06-17 16:26:12.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-06-17 16:26:12.000000000","tz":-420},"subject":"Install virtualenv in proposal jobs","message":"Install virtualenv in proposal jobs\n\nWe need this to use virtualenv -p as venv doesn\u0027t support venvs for\ndifferent python versions.\n\nChange-Id: I0456da209310e1b9d0af6f1281433b3d9edb7296\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4fcc8adcf61c9906bca861cff5b5f0e83a801668"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4fcc8adcf61c9906bca861cff5b5f0e83a801668"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
