)]}'
{"id":"openstack%2Fproject-config~846899","triplet_id":"openstack%2Fproject-config~master~If568eafcce0341046ba512532d92d9ad75ba40e4","project":"openstack/project-config","branch":"master","topic":"add-cinder-storwize","attention_set":{},"removed_from_attention_set":{"34352":{"account":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"last_update":"2022-06-22 23:39:03.000000000","reason":"Change was submitted"},"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-06-22 23:39:03.000000000","reason":"Change was submitted"},"20870":{"account":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"last_update":"2022-06-22 23:39:03.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-06-22 23:39:03.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-06-22 23:27:47.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If568eafcce0341046ba512532d92d9ad75ba40e4","subject":"Add Cinder IBM Storwize charm","status":"MERGED","created":"2022-06-21 04:44:33.000000000","updated":"2022-06-22 23:59:51.000000000","submitted":"2022-06-22 23:39:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846899-add-cinder-storwize","meta_rev_id":"4ae4c34d1c013d25c92c1399619aae4ded3a6cfa","_number":846899,"virtual_id_number":846899,"owner":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-22 23:39:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-06-22 23:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2022-06-22 23:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2022-06-22 23:27:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-21 04:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-21 11:21:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-06-21 20:47:09.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-06-22 06:42:00.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"},{"updated":"2022-06-22 23:09:16.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"f3fb3425fbe1b72af3fbfd471f62296339e516cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"date":"2022-06-21 04:44:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddab7e8a5e0c6b948c3edb9a1ca4490afdecc339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 04:52:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/12f0f5076d5b4597b62fb644052bc146 : SUCCESS in 7m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/42d4420699fe4f4f84887d876a71cdb9 : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a1d86fd87e6ef20dc376c27c8d50246a91e16b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"date":"2022-06-21 05:04:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7a9154d36ad111eed853b150eaa457f8b0570028","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"date":"2022-06-21 05:08:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7081c397e6e5816f85ad12baca3e13e634436704","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 05:15:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9743f861c4d942b484d38f3d15373c2c : SUCCESS in 5m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6e891bd87a8d443abba3772d8e5d1133 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"3f6d870b9d09066a86535f083fbfb9f4ce885117","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-06-21 11:21:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"80adf7bc6df957ec4c265491504ad282a7b5afec","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-06-21 20:47:09.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f0583790ef03a87ab89abe3fe6d2614f5c98993","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-06-22 06:42:00.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f60614be4b6af2b0513fe62620b6fa4e8195756","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"date":"2022-06-22 09:13:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"38bc5b6e293455c7ccbd8342b4400e4a0bba9e66","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-06-22 09:14:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"af2cb9211ebb5762cf660b48a46e7605f8cb79dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"date":"2022-06-22 23:07:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dfab8d0b057143eda8e55f331161ba0640220399","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-06-22 23:09:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0858194c18c431b37c3fba2ac606e02d5a82f9a9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"date":"2022-06-22 23:09:48.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"63fcebd15ec5714a68cdfffe2d9432966c4f5574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:16:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0995fafecd634020a068e607ce793ad6 : SUCCESS in 6m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e512773269f748ee816a28e2e1d688f8 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":4},{"id":"e33f714ed098b8b61c2b30559c551433cd00f790","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-06-22 23:27:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"46ded0a5cc022592b75ae7f2219b41a055db8965","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:28:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1c102d6dba04bdcf03d97aa465a069d002472d89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:39:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4630413b002d4c799f1a25ecc19b621b : SUCCESS in 6m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/88de6d85c7144c719a11b7b76a778fda : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":4},{"id":"3b52ac5ec53fa3186dc5a69f5c89aead2c2526e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:39:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4ae4c34d1c013d25c92c1399619aae4ded3a6cfa","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 23:59:51.000000000","message":"Patch Set 4:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/439b374ba5944703b2d15b7c51e6d3dc : SUCCESS in 7m 21s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/bf6b4aaf087f4b3e847b9ba6485b8146 : FAILURE in 9m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"70658e37f3bed35f2ea254bd9984b6849e7e9a8f","revisions":{"793306e857ad0642ca1a416a46ad1fff8f6ceeeb":{"kind":"REWORK","_number":1,"created":"2022-06-21 04:44:33.000000000","uploader":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"ref":"refs/changes/99/846899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/846899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/846899/1"}}},"commit":{"parents":[{"commit":"be4b2767297cefaedb9c422fed5930a71f31c251","subject":"Only install python3.8-venv in translation update jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be4b2767297cefaedb9c422fed5930a71f31c251"}]}],"author":{"name":"Samuel Walladge","email":"samuel.walladge@canonical.com","date":"2022-06-21 04:44:13.000000000","tz":570},"committer":{"name":"Samuel Walladge","email":"samuel.walladge@canonical.com","date":"2022-06-21 04:44:13.000000000","tz":570},"subject":"Add cinder ibm storwize charm","message":"Add cinder ibm storwize charm\n\nChange-Id: If568eafcce0341046ba512532d92d9ad75ba40e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/793306e857ad0642ca1a416a46ad1fff8f6ceeeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/793306e857ad0642ca1a416a46ad1fff8f6ceeeb"}]},"branch":"refs/heads/master"},"d8270505116c84e8e0ba1fb99780d6b98b3ecea1":{"kind":"REWORK","_number":2,"created":"2022-06-21 05:04:44.000000000","uploader":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"ref":"refs/changes/99/846899/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/846899/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/846899/2"}}},"commit":{"parents":[{"commit":"be4b2767297cefaedb9c422fed5930a71f31c251","subject":"Only install python3.8-venv in translation update jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be4b2767297cefaedb9c422fed5930a71f31c251"}]}],"author":{"name":"Samuel Walladge","email":"samuel.walladge@canonical.com","date":"2022-06-21 04:44:13.000000000","tz":570},"committer":{"name":"Samuel Walladge","email":"samuel.walladge@canonical.com","date":"2022-06-21 05:04:21.000000000","tz":570},"subject":"Add Cinder IBM Storwize charm","message":"Add Cinder IBM Storwize charm\n\nChange-Id: If568eafcce0341046ba512532d92d9ad75ba40e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d8270505116c84e8e0ba1fb99780d6b98b3ecea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d8270505116c84e8e0ba1fb99780d6b98b3ecea1"}]},"branch":"refs/heads/master"},"e32d8b951196f6937217686b408e0b88b67d466c":{"kind":"REWORK","_number":3,"created":"2022-06-21 05:08:37.000000000","uploader":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"ref":"refs/changes/99/846899/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/846899/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/846899/3"}}},"commit":{"parents":[{"commit":"be4b2767297cefaedb9c422fed5930a71f31c251","subject":"Only install python3.8-venv in translation update jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be4b2767297cefaedb9c422fed5930a71f31c251"}]}],"author":{"name":"Samuel Walladge","email":"samuel.walladge@canonical.com","date":"2022-06-21 04:44:13.000000000","tz":570},"committer":{"name":"Samuel Walladge","email":"samuel.walladge@canonical.com","date":"2022-06-21 05:08:27.000000000","tz":570},"subject":"Add Cinder IBM Storwize charm","message":"Add Cinder IBM Storwize charm\n\nChange-Id: If568eafcce0341046ba512532d92d9ad75ba40e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e32d8b951196f6937217686b408e0b88b67d466c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e32d8b951196f6937217686b408e0b88b67d466c"}]},"branch":"refs/heads/master"},"70658e37f3bed35f2ea254bd9984b6849e7e9a8f":{"kind":"REWORK","_number":4,"created":"2022-06-22 23:07:54.000000000","uploader":{"_account_id":34352,"name":"Samuel Allan","email":"samuel.allan@canonical.com","username":"swalladge","status":"@Canonical"},"ref":"refs/changes/99/846899/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/99/846899/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/99/846899/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/99/846899/4"}}},"commit":{"parents":[{"commit":"be4b2767297cefaedb9c422fed5930a71f31c251","subject":"Only install python3.8-venv in translation update jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be4b2767297cefaedb9c422fed5930a71f31c251"}]}],"author":{"name":"Samuel Walladge","email":"samuel.walladge@canonical.com","date":"2022-06-21 04:44:13.000000000","tz":570},"committer":{"name":"Samuel Walladge","email":"samuel.walladge@canonical.com","date":"2022-06-22 22:56:11.000000000","tz":570},"subject":"Add Cinder IBM Storwize charm","message":"Add Cinder IBM Storwize charm\n\nChange-Id: If568eafcce0341046ba512532d92d9ad75ba40e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70658e37f3bed35f2ea254bd9984b6849e7e9a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70658e37f3bed35f2ea254bd9984b6849e7e9a8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
