)]}'
{"id":"openstack%2Fproject-config~847135","triplet_id":"openstack%2Fproject-config~master~I554f8892a7e19d609e9de907143168e8d17fb6ab","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-02-21 12:49:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I554f8892a7e19d609e9de907143168e8d17fb6ab","subject":"Add hashicorp-vault-plugin-secrets project","status":"ABANDONED","created":"2022-06-22 08:48:55.000000000","updated":"2024-02-21 13:47:01.000000000","total_comment_count":9,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"801e5e3996d3b04b6c52b59aa050bf67fb37ea59","_number":847135,"virtual_id_number":847135,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-09 08:29:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-22 08:57:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-24 19:28:15.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-11-09 08:56:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"55dd541acd14a571128994e2eafb3c2cc23909bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-06-22 08:48:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c0011abd6c18c2b1abb8ddf6a888bfa9c1118e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 08:57:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/522b02a1d69b4084b0c143bc2f027232 : SUCCESS in 6m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6adfdc39e6ca478cab246dd9b9300b63 : FAILURE in 3m 47s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/4d3d86d220ad4c7d81ebb5c41c1ee34c : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"42da18fb22413b20f45b79f02b4c0f4770ba7815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-06-22 09:05:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82952593cdc66b4ff57eb71d1022238262d981d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 09:14:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3c10978c19294035a6eea8e4d6b336a9 : SUCCESS in 7m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3ff34fc7bf2f4a68996315648e2c965d : SUCCESS in 3m 46s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/bebfc86b3f6542858fda4909146c6b44 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"5f7016ba39b884487b552a30d0de0176fa9ba664","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-24 19:28:15.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nIf this is going to be in the openstack namespace, TC policy is that it must be associated with a project team or SIG. For that, you\u0027ll need to also propose a corresponding openstack/governance change and link it in the commit message or a review comment so we\u0027re able to confirm it\u0027s sanctioned by the OpenStack project.\n\nIf you don\u0027t mean for it to be part of OpenStack in an official sense (the commit message seems to imply you don\u0027t?) then you can propose it to a different new or existing namespace as appropriate.","accounts_in_message":[],"_revision_number":2},{"id":"2301ae80ec1496cfccc2688fe306575ec01e38c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-11-09 08:17:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3b3f6b1dec21801f07f29c1b0a11c6bfbff9a90b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-11-09 08:18:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b50c52ed8ada2cf31f831cff5f170897f32ba054","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-11-09 08:19:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fa9b44efa28bf51fa1405802867e28c12d543cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 08:29:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/75be7f1ebb7f49dabb8c5077c3a7e515\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e4bd6de0da0e4aad8c14dc93f26e2533 : SUCCESS in 5m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b3ad48659610442d923c0213456c7048 : FAILURE in 3m 15s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/5794f3b61a5449008f0bfdd611c15fc6 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":4},{"id":"8fc7860f7b6812ebf36e78e583038015d1ab9ad2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-09 08:56:46.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"16f99bee75183e939ef7aad0c752303f5a6c5b78","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-11-09 13:58:48.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2a8e7cf1fe12be46fddfe4dfd7c64b0a09b43e47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-02-21 12:49:28.000000000","message":"Abandoned\n\nHashiCorp license seems to be major issue","accounts_in_message":[],"_revision_number":4},{"id":"801e5e3996d3b04b6c52b59aa050bf67fb37ea59","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2024-02-21 13:47:01.000000000","message":"Patch Set 4:\n\nArtem Goncharov wrote:\n\n\u003e HashiCorp license seems to be major issue\n\nWould basing it on https://github.com/openbao/openbao (open source fork of Vault) have been an option?","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fbbc5352161c952394145f9c248a4afab812ca65","revisions":{"5a6efdbe2e38547401e769cd3a0e56912a4375a0":{"kind":"REWORK","_number":1,"created":"2022-06-22 08:48:55.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/35/847135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/847135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/847135/1"}}},"commit":{"parents":[{"commit":"feb7b7f98ef79624bdd6f7eb9553e6599bc171b9","subject":"Merge \"Re-maintenance neutron-fwaas*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/feb7b7f98ef79624bdd6f7eb9553e6599bc171b9"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-06-22 08:44:23.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-06-22 08:44:23.000000000","tz":120},"subject":"Add hashicorp-vault-plugin-secrets project","message":"Add hashicorp-vault-plugin-secrets project\n\nMove\nhttps://github.com/opentelekomcloud/vault-plugin-secrets-openstack\nproject to the opendev. For now this is not meant to be official\nproject. The project is currently maintained by 3rd party Zuul,\ntherefore do not import it directly.\n\nChange-Id: I554f8892a7e19d609e9de907143168e8d17fb6ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a6efdbe2e38547401e769cd3a0e56912a4375a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a6efdbe2e38547401e769cd3a0e56912a4375a0"}]},"branch":"refs/heads/master"},"b5de391f9a8f0fd285be17f2bc5647aaad42837f":{"kind":"REWORK","_number":2,"created":"2022-06-22 09:05:31.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/35/847135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/847135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/847135/2"}}},"commit":{"parents":[{"commit":"feb7b7f98ef79624bdd6f7eb9553e6599bc171b9","subject":"Merge \"Re-maintenance neutron-fwaas*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/feb7b7f98ef79624bdd6f7eb9553e6599bc171b9"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-06-22 08:44:23.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-06-22 09:05:25.000000000","tz":120},"subject":"Add hashicorp-vault-plugin-secrets project","message":"Add hashicorp-vault-plugin-secrets project\n\nMove\nhttps://github.com/opentelekomcloud/vault-plugin-secrets-openstack\nproject to the opendev. For now this is not meant to be official\nproject. The project is currently maintained by 3rd party Zuul,\ntherefore do not import it directly.\n\nChange-Id: I554f8892a7e19d609e9de907143168e8d17fb6ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b5de391f9a8f0fd285be17f2bc5647aaad42837f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b5de391f9a8f0fd285be17f2bc5647aaad42837f"}]},"branch":"refs/heads/master"},"456274c089ac89e87aebf4534fb6cbaf5e1be471":{"kind":"REWORK","_number":3,"created":"2023-11-09 08:17:06.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/35/847135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/847135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/847135/3"}}},"commit":{"parents":[{"commit":"feb7b7f98ef79624bdd6f7eb9553e6599bc171b9","subject":"Merge \"Re-maintenance neutron-fwaas*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/feb7b7f98ef79624bdd6f7eb9553e6599bc171b9"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-06-22 08:44:23.000000000","tz":120},"committer":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2023-11-09 08:16:25.000000000","tz":60},"subject":"Add hashicorp-vault-plugin-secrets project","message":"Add hashicorp-vault-plugin-secrets project\n\nMove\nhttps://github.com/opentelekomcloud/vault-plugin-secrets-openstack\nproject to the opendev. For now this is not meant to be official\nproject. The project is currently maintained by 3rd party Zuul,\ntherefore do not import it directly.\n\nChange-Id: I554f8892a7e19d609e9de907143168e8d17fb6ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/456274c089ac89e87aebf4534fb6cbaf5e1be471"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/456274c089ac89e87aebf4534fb6cbaf5e1be471"}]},"branch":"refs/heads/master"},"fbbc5352161c952394145f9c248a4afab812ca65":{"kind":"REWORK","_number":4,"created":"2023-11-09 08:18:06.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/35/847135/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/847135/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/847135/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/847135/4"}}},"commit":{"parents":[{"commit":"feb7b7f98ef79624bdd6f7eb9553e6599bc171b9","subject":"Merge \"Re-maintenance neutron-fwaas*\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/feb7b7f98ef79624bdd6f7eb9553e6599bc171b9"}]}],"author":{"name":"Artem Goncharov","email":"Artem.goncharov@gmail.com","date":"2022-06-22 08:44:23.000000000","tz":120},"committer":{"name":"gtema","email":"artem.goncharov@gmail.com","date":"2023-11-09 08:17:57.000000000","tz":60},"subject":"Add hashicorp-vault-plugin-secrets project","message":"Add hashicorp-vault-plugin-secrets project\n\nMove\nhttps://github.com/opentelekomcloud/vault-plugin-secrets-openstack\nproject to the opendev. For now this is not meant to be official\nproject. The project is currently maintained by 3rd party Zuul,\ntherefore do not import it directly.\n\nChange-Id: I554f8892a7e19d609e9de907143168e8d17fb6ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fbbc5352161c952394145f9c248a4afab812ca65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fbbc5352161c952394145f9c248a4afab812ca65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
