)]}'
{"id":"openstack%2Fproject-config~847414","triplet_id":"openstack%2Fproject-config~master~I8299600120273bbd6619574081fe0212cfa4f2a5","project":"openstack/project-config","branch":"master","topic":"retire-osh-deployments","attention_set":{},"removed_from_attention_set":{"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-06-27 23:05:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8299600120273bbd6619574081fe0212cfa4f2a5","subject":"Retire openstack-helm-deployments repo","status":"MERGED","created":"2022-06-23 21:38:07.000000000","updated":"2022-06-27 23:56:45.000000000","submitted":"2022-06-27 23:05:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847414-retire-osh-deployments","meta_rev_id":"d945410d2caffdc7bfbc1228092b9e7cf907c464","_number":847414,"virtual_id_number":847414,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-27 23:05:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2022-06-27 22:25:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-24 23:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-27 22:55:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-23 21:51:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 23:55:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-06-27 22:25:36.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"a45a63d0b5a8931e61ef43e771dac0a290b6e9f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-06-23 21:38:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60f9015a2635ec1d4175d14a06dc6f8fbfe8ecb4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-06-23 21:39:26.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"c3368dd6a34c9b1653d7f4ec5f38788ea5caa6bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 21:51:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5216db9bfb0748c18eb73c03d2ae41ef : SUCCESS in 7m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/858ba4c82f0847d69c74a35d8111273b : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2dbe0421d5b83355996ca475dede6ab94dc22888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-06-23 22:03:39.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"da634ef64ba08b7379d127b4163c90c744e44f02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 22:13:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e06025985b3647b99a5e40405103b86f : SUCCESS in 7m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/da87a61aede040caaa51118353144de2 : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"43cf9cd145727a2f65c7375956537347bebdb72f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-06-24 21:46:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0c822c54faef14affdd7e0f29119dde8af6bd673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-06-24 21:47:11.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"ec720e8793609c57c5de1e56fd80c1c7bb796ec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-06-24 21:56:26.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a1578247e044fed58085e8bfdc7884a795695aa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-06-24 21:56:58.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"c70e117aa6539d6bbcff1514e5f7633390b8dacd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 22:10:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6aeed22f47f3457ca1a38367410905c4 : SUCCESS in 7m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b6dcce8b94f941b4a8360238c09b0368 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":6},{"id":"a11e677e0a2bf9237aa82d359e691bd17637a2a6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-06-24 23:55:26.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b1e9e315449713496d0f3f7958aaa9abbb57e215","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-06-27 22:25:36.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"73da1d52940e4baf87900fba7175468522f64f70","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-06-27 22:55:12.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"afa8f175bc79f166db649837c3d8a2efa4ff27d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 22:56:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8f39bab7118b4fd6d480c115ea14e55b87df9a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 23:05:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0aee387a7c1942a4b528a467ccceefa5 : SUCCESS in 7m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/26ee755fe0f64836bd553c24c6051e28 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":6},{"id":"769cbb843db20a2864e998abcc6edd2e8ae49450","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 23:05:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d945410d2caffdc7bfbc1228092b9e7cf907c464","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 23:56:45.000000000","message":"Patch Set 6:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/84b10264ee5248f29061c70b07d095ea : SUCCESS in 7m 15s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/0122c57755974859882f9ec6e84f2294 : FAILURE in 8m 50s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"82b35d557fc8b981759b137ed50a73d2faec7d21","revisions":{"f7f683922d549ef0acad201386d6cd23f982ba2c":{"kind":"REWORK","_number":1,"created":"2022-06-23 21:38:07.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/847414/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/847414/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/847414/1"}}},"commit":{"parents":[{"commit":"11772c4ea0abf2e9f425cc28f4bd03b2f85b873e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/11772c4ea0abf2e9f425cc28f4bd03b2f85b873e"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-23 21:36:54.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-23 21:36:54.000000000","tz":-300},"subject":"Retire openstack-helm-deployments repo","message":"Retire openstack-helm-deployments repo\n\nThis change is part of the process to retire the\nopenstack-helm-deployments repository.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/847413\nDepends-on: https://review.opendev.org/c/openstack/openstack-helm-deployments/+/847410\nChange-Id: I8299600120273bbd6619574081fe0212cfa4f2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7f683922d549ef0acad201386d6cd23f982ba2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7f683922d549ef0acad201386d6cd23f982ba2c"}]},"branch":"refs/heads/master"},"0269615f14769496f22716df32f9d7de1156be6f":{"kind":"REWORK","_number":2,"created":"2022-06-23 22:03:39.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/847414/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/847414/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/847414/2"}}},"commit":{"parents":[{"commit":"11772c4ea0abf2e9f425cc28f4bd03b2f85b873e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/11772c4ea0abf2e9f425cc28f4bd03b2f85b873e"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-23 21:36:54.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-23 22:03:38.000000000","tz":0},"subject":"Retire openstack-helm-deployments repo","message":"Retire openstack-helm-deployments repo\n\nThis change is part of the process to retire the\nopenstack-helm-deployments repository.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/847413\nDepends-on: https://review.opendev.org/c/openstack/openstack-helm-deployments/+/847410\nChange-Id: I8299600120273bbd6619574081fe0212cfa4f2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0269615f14769496f22716df32f9d7de1156be6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0269615f14769496f22716df32f9d7de1156be6f"}]},"branch":"refs/heads/master"},"155e866c09859e6b63cdfc18de01b5cc3251089e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-06-24 21:46:26.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/847414/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/847414/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/847414/3"}}},"commit":{"parents":[{"commit":"00bd5925718ab2a6dd299b0d2cad6bfe9eb63194","subject":"End project gating for openstack-helm deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00bd5925718ab2a6dd299b0d2cad6bfe9eb63194"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-23 21:36:54.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-24 21:46:26.000000000","tz":0},"subject":"Retire openstack-helm-deployments repo","message":"Retire openstack-helm-deployments repo\n\nThis change is part of the process to retire the\nopenstack-helm-deployments repository.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/847413\nDepends-on: https://review.opendev.org/c/openstack/openstack-helm-deployments/+/847410\nChange-Id: I8299600120273bbd6619574081fe0212cfa4f2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/155e866c09859e6b63cdfc18de01b5cc3251089e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/155e866c09859e6b63cdfc18de01b5cc3251089e"}]},"branch":"refs/heads/master","description":"Rebase"},"00854d73d5d9aa49581c3c8dc9a4beff57d26661":{"kind":"REWORK","_number":4,"created":"2022-06-24 21:47:11.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/847414/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/847414/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/847414/4"}}},"commit":{"parents":[{"commit":"00bd5925718ab2a6dd299b0d2cad6bfe9eb63194","subject":"End project gating for openstack-helm deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00bd5925718ab2a6dd299b0d2cad6bfe9eb63194"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-23 21:36:54.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-24 21:47:08.000000000","tz":0},"subject":"Retire openstack-helm-deployments repo","message":"Retire openstack-helm-deployments repo\n\nThis change is part of the process to retire the\nopenstack-helm-deployments repository.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/847413\nDepends-on: https://review.opendev.org/c/openstack/openstack-helm-deployments/+/847410\nChange-Id: I8299600120273bbd6619574081fe0212cfa4f2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00854d73d5d9aa49581c3c8dc9a4beff57d26661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/00854d73d5d9aa49581c3c8dc9a4beff57d26661"}]},"branch":"refs/heads/master"},"c44b095c951fdc06ee8cfdc632b75554d77b0896":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-06-24 21:56:26.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/847414/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/847414/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/847414/5"}}},"commit":{"parents":[{"commit":"fc8e2658ec46a7e228229b5536841c2f1484ff3c","subject":"End project gating for openstack-helm deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fc8e2658ec46a7e228229b5536841c2f1484ff3c"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-23 21:36:54.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-24 21:56:26.000000000","tz":0},"subject":"Retire openstack-helm-deployments repo","message":"Retire openstack-helm-deployments repo\n\nThis change is part of the process to retire the\nopenstack-helm-deployments repository.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/847413\nDepends-on: https://review.opendev.org/c/openstack/openstack-helm-deployments/+/847410\nChange-Id: I8299600120273bbd6619574081fe0212cfa4f2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c44b095c951fdc06ee8cfdc632b75554d77b0896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c44b095c951fdc06ee8cfdc632b75554d77b0896"}]},"branch":"refs/heads/master","description":"Rebase"},"82b35d557fc8b981759b137ed50a73d2faec7d21":{"kind":"REWORK","_number":6,"created":"2022-06-24 21:56:58.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/14/847414/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/847414/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/847414/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/847414/6"}}},"commit":{"parents":[{"commit":"fc8e2658ec46a7e228229b5536841c2f1484ff3c","subject":"End project gating for openstack-helm deployments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fc8e2658ec46a7e228229b5536841c2f1484ff3c"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-23 21:36:54.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-06-24 21:56:57.000000000","tz":0},"subject":"Retire openstack-helm-deployments repo","message":"Retire openstack-helm-deployments repo\n\nThis change is part of the process to retire the\nopenstack-helm-deployments repository.\n\nDepends-on: https://review.opendev.org/c/openstack/governance/+/847413\nDepends-on: https://review.opendev.org/c/openstack/openstack-helm-deployments/+/847410\nChange-Id: I8299600120273bbd6619574081fe0212cfa4f2a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82b35d557fc8b981759b137ed50a73d2faec7d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82b35d557fc8b981759b137ed50a73d2faec7d21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
