)]}'
{"id":"openstack%2Fproject-config~847973","triplet_id":"openstack%2Fproject-config~master~I1c4705ccf92b2d4c6ca372b26572e674d93e1bdb","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I1c4705ccf92b2d4c6ca372b26572e674d93e1bdb","subject":"Update openstack-ansible grafana dashboard","status":"MERGED","created":"2022-06-28 13:31:39.000000000","updated":"2022-07-03 23:45:05.000000000","submitted":"2022-07-03 23:26:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847973","meta_rev_id":"b647cdca759a91139c8d1149db8943ee71e9c39c","_number":847973,"virtual_id_number":847973,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-03 23:26:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-28 20:51:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-03 23:15:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 13:52:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-28 20:51:34.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"f13064d4411a362bd2a3d1b234ead112e9c53e3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2022-06-28 13:31:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89b06c3a82ade320365295794d680b5663da6df8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:52:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0afbcfc616c943639cbee11fdb6fd86a : SUCCESS in 18m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a08a93af0fd04fa7a01c177900a4a0d9 : SUCCESS in 6m 33s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/d613008f7618481e960d232270e160b0 : SUCCESS in 4m 02s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/d46a282c911442199599bb551e43e439 : SUCCESS in 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"281c88495ef687027241bb57b85e1a6b8f9cb9d8","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-28 20:51:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67c749bd4ad0791f9eb594642b35a2ce7a628563","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-03 23:15:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff468d75037b0bfa79b7a8cfe4d46e7b9e71a1b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 23:15:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98b6985824e2c29d4cfd1c48ea86c84203ef9caf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 23:26:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7481834d76f9427e8043b5befe959848\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8d6db8389efa406a89808d814a019ac9 : SUCCESS in 5m 48s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/1f6305b50d804992a7410ccea7a432a3 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b724daab4ed9b3e56d02313597271c055b3f9da1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 23:26:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b647cdca759a91139c8d1149db8943ee71e9c39c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 23:45:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95366b684fce4ca09affa1f97309fe99\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/7709a26034eb42679ad2b65dad478978 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3b01749949d7e6c553fc31fec9e12f6e3dea8df","revisions":{"a3b01749949d7e6c553fc31fec9e12f6e3dea8df":{"kind":"REWORK","_number":1,"created":"2022-06-28 13:31:39.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/73/847973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/847973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/847973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/847973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/847973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/847973/1"}}},"commit":{"parents":[{"commit":"418ea119713eb78517882368373298f496ffeec7","subject":"Merge \"project-config-grafana: add opendev-buildset-registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/418ea119713eb78517882368373298f496ffeec7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-06-28 13:28:38.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2022-06-28 13:28:38.000000000","tz":60},"subject":"Update openstack-ansible grafana dashboard","message":"Update openstack-ansible grafana dashboard\n\nUpdate with the current set of operating systems running jobs on\nmaster branch:\n\nRemove Centos 7\nSwitch Centos 8 to Centos 8 stream\nAdd Centos 9 Stream\nRemove Ubuntu Bionic\nAdd Ubuntu Jammy\nAdd Rockylinux 8\n\nChange-Id: I1c4705ccf92b2d4c6ca372b26572e674d93e1bdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a3b01749949d7e6c553fc31fec9e12f6e3dea8df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a3b01749949d7e6c553fc31fec9e12f6e3dea8df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
