)]}'
{"id":"openstack%2Fproject-config~848034","triplet_id":"openstack%2Fproject-config~master~I35a6093229951659d7b80015275aebc386b4cd95","project":"openstack/project-config","branch":"master","topic":"zuul-tenant-cleanup","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-06-30 17:28:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I35a6093229951659d7b80015275aebc386b4cd95","subject":"Remove x/neutron-classifier from Zuul tenant config","status":"MERGED","created":"2022-06-28 21:22:34.000000000","updated":"2022-06-30 17:52:27.000000000","submitted":"2022-06-30 17:28:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848034-zuul-tenant-cleanup","meta_rev_id":"9fa760d3b9f620709a7d0b5500d99bc3474d3898","_number":848034,"virtual_id_number":848034,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-30 17:28:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-06-29 15:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-30 17:16:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-30 17:16:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 21:36:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-29 15:53:30.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-06-30 17:16:54.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"57f9a371ef8e1ce819b4daf2cc2a2535c957078e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-06-28 21:22:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b1202f097454226607168a399a08773f6be74f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 21:36:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f20e19065f3a4688ac7baaa87407d732 : SUCCESS in 4m 34s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5fed3f3b68214ac0be5b754194ffa851 : SUCCESS in 8m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/88f4cdb5bd71493abc787f2ed036e565 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"80ce8ae535843889788c064cce2d6492f43a7c7c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-29 15:53:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0202260e323404ff6f801d1be19ca26c26703864","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-06-30 17:16:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f9a0918165ff387cf3f9039759110bce163f63f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 17:17:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c883abda5275de64a98503f128f161375e7b9129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 17:28:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9299000e28104b53a35e0e5b1bb8d558 : SUCCESS in 7m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/628f21e0a4ac490c9dcb6b58a9a62efa : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"81a6533fcbbf5202540856026d38bbabe6f0a7b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 17:28:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9fa760d3b9f620709a7d0b5500d99bc3474d3898","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 17:52:27.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/277336b3f6b24e99b8395a0ed3372ec4 : FAILURE in 8m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d806f27eb3f8b9778cff4c42cdaa2f5bbd6de12","revisions":{"8d806f27eb3f8b9778cff4c42cdaa2f5bbd6de12":{"kind":"REWORK","_number":1,"created":"2022-06-28 21:22:34.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/34/848034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/848034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/848034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/848034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/848034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/848034/1"}}},"commit":{"parents":[{"commit":"598407a05167507b9719ab63bca38dde8984aa73","subject":"Remove windmill from zuul tenant config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/598407a05167507b9719ab63bca38dde8984aa73"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-06-28 21:20:47.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-06-28 21:20:47.000000000","tz":-420},"subject":"Remove x/neutron-classifier from Zuul tenant config","message":"Remove x/neutron-classifier from Zuul tenant config\n\nThis repo uses deprecated per pipeline Zuul queue: configuration. I\npushed a change [0] to correct this, but the project seems to be idle.\nThe last merge occured at the end of 2019. We remove this repo from the\nZuul tenant config to avoid errors when Zuul stops supporting this\ndeprecated syntax.\n\n[0] https://review.opendev.org/c/x/neutron-classifier/+/842623\n\nChange-Id: I35a6093229951659d7b80015275aebc386b4cd95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d806f27eb3f8b9778cff4c42cdaa2f5bbd6de12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8d806f27eb3f8b9778cff4c42cdaa2f5bbd6de12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
