)]}'
{"id":"openstack%2Fproject-config~848212","triplet_id":"openstack%2Fproject-config~master~I74822881251c627d8e783438fa82e39aeca5645e","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-08-04 00:11:09.000000000","reason":"Change was submitted"},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-08-03 06:37:19.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I74822881251c627d8e783438fa82e39aeca5645e","subject":"grafana: update nodepool-dib status page","status":"MERGED","created":"2022-06-30 04:37:40.000000000","updated":"2022-08-04 00:54:58.000000000","submitted":"2022-08-04 00:11:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"848212","meta_rev_id":"c8613203c438542efb1ea0ffc9248f8971c22eb5","_number":848212,"virtual_id_number":848212,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 00:11:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-07-22 22:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-04 00:02:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-30 04:47:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-22 22:12:04.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"29bf06b48591154e0f190d17569754df03c3736d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-30 04:37:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a590eb904448d06bc8654af2853dbb401e3ee587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 04:47:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f79ac2b5fe634128ad819a2d0fc177f4 : SUCCESS in 8m 27s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e88c91f88ff442c88c74e45900672de0 : SUCCESS in 7m 39s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/f8845b97813e4b9b98bd80614df2ac52 : FAILURE in 4m 08s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/ed35a4b2c8ea41f4bb5dc70fde06e866 : FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9be80f8ded84605202c20c3b360fd0e6a24d4d91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-30 04:51:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7862b1118d82b75868b6098ff9133c4f74f85fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 05:08:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9de81af7c4af4680ba50b5e3d80188bc : SUCCESS in 15m 48s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/15af1028b9c14d2392d5f8f4ee0e88d1 : SUCCESS in 6m 47s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/f9ac0e8f0de9416b917fac3cca60c068 : SUCCESS in 3m 54s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/a36c3ac08ae849b19db8ca9c0e38f30d : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":2},{"id":"9452acdc336ff7c2f4659340dbbe5251dccacd5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-06-30 05:53:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"95d988f94d99925ef46d7f3821a2f1d46cd9c3e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 06:13:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/14da75e4a1e14a59b71ea70f200db8ef : SUCCESS in 18m 02s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/41ad63269c3a4189986800c4e2e2764f : SUCCESS in 7m 09s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/ceafdc29e5334cd1b272e961b6addc72 : SUCCESS in 3m 42s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/24fe2361d9724c7ea13d5bad848f8a8a : SUCCESS in 12m 59s","accounts_in_message":[],"_revision_number":3},{"id":"9bc66b0c6fe346be3db3543e33e08f6b47f5ac10","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-07-22 22:12:04.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eac51e19f4dcde5a50c1a9e6cba5d7361fac3d7a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-03 06:37:19.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"db2613799c5c7ed8284108ec5fc817d0fd8120bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 06:55:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b55c2d372134bfd90d9820f9d9b9555\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/44884d98f355442dba1be1c756483f48 : SUCCESS in 17m 55s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7278d01c45154a409207ae19fbdc6095 : SUCCESS in 7m 20s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/a673219037c243d1931490a98764fa4b : SUCCESS in 3m 16s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/5ac295116f744ef59e4d12124be83238 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":3},{"id":"7c67b2b51acc3c999cb04e740315f66708f34875","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-04 00:02:19.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7c0d23374230f53d83976769fae67ff59cf118e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 00:02:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2c4a5f32d7307bf7d28bb927865854140ff8a10d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 00:11:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52022964cdea49e19dee6dc3e8f00f20\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6293bd6e678841d6beceb2f730ea3d74 : SUCCESS in 6m 10s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/ab05994f6c1346389e77537199ce3e12 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"f45a52e1ea26cb8b721a7aeb22a60e4a61bfc1b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 00:11:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"c8613203c438542efb1ea0ffc9248f8971c22eb5","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 00:54:58.000000000","message":"Patch Set 3:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe327a7bff7e45348e391ba47d7adbdb\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/d7796fda087f4bf3ae7c781a4ecb3ae9 : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5c57b07f96cfcf6d302c203b53453fe2f5e3cb9a","revisions":{"ebe665985b70bfb3b59edef9a90450c2152b02b7":{"kind":"REWORK","_number":1,"created":"2022-06-30 04:37:40.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/12/848212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/848212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/848212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/848212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/848212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/848212/1"}}},"commit":{"parents":[{"commit":"859005d497fe14bcd7335bfd8bd95a0bdaa5e63b","subject":"grafana: add helper script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/859005d497fe14bcd7335bfd8bd95a0bdaa5e63b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-30 04:35:06.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-30 04:37:23.000000000","tz":600},"subject":"[wip] update nodepool dib graph","message":"[wip] update nodepool dib graph\n\nChange-Id: I74822881251c627d8e783438fa82e39aeca5645e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebe665985b70bfb3b59edef9a90450c2152b02b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebe665985b70bfb3b59edef9a90450c2152b02b7"}]},"branch":"refs/heads/master"},"de4bce770570207db92693f78a5e040bee8274bd":{"kind":"REWORK","_number":2,"created":"2022-06-30 04:51:45.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/12/848212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/848212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/848212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/848212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/848212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/848212/2"}}},"commit":{"parents":[{"commit":"859005d497fe14bcd7335bfd8bd95a0bdaa5e63b","subject":"grafana: add helper script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/859005d497fe14bcd7335bfd8bd95a0bdaa5e63b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-30 04:35:06.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-30 04:51:29.000000000","tz":600},"subject":"[wip] update nodepool dib graph","message":"[wip] update nodepool dib graph\n\nChange-Id: I74822881251c627d8e783438fa82e39aeca5645e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de4bce770570207db92693f78a5e040bee8274bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de4bce770570207db92693f78a5e040bee8274bd"}]},"branch":"refs/heads/master"},"5c57b07f96cfcf6d302c203b53453fe2f5e3cb9a":{"kind":"REWORK","_number":3,"created":"2022-06-30 05:53:53.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/12/848212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/848212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/848212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/848212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/848212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/848212/3"}}},"commit":{"parents":[{"commit":"859005d497fe14bcd7335bfd8bd95a0bdaa5e63b","subject":"grafana: add helper script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/859005d497fe14bcd7335bfd8bd95a0bdaa5e63b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-30 04:35:06.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-06-30 05:50:41.000000000","tz":600},"subject":"grafana: update nodepool-dib status page","message":"grafana: update nodepool-dib status page\n\nUpdate this to new panel types and modify layout a bit.  This uses a\nvariable to map over the stats, rather than the old template model.\n\nThis replaces the non-working thresholds that show the \"OK\"/\"FAILED\"\nwith value mappings which should be more reliable.\n\nThis also includes a heatmap at the top -- it\u0027s a bit weird because\nit\u0027s mapping the return values of the builds, so can range between\n0-\u003e128.  The return code value doesn\u0027t really help diagnose anything,\nbut it can highlight if things are failing in the same way (i.e. with\nthe same return code).\n\nChange-Id: I74822881251c627d8e783438fa82e39aeca5645e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5c57b07f96cfcf6d302c203b53453fe2f5e3cb9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5c57b07f96cfcf6d302c203b53453fe2f5e3cb9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
