)]}'
{"id":"openstack%2Fproject-config~848800","triplet_id":"openstack%2Fproject-config~master~I22c549b5f9b9e3882fcd2340946d2850b0b2f86b","project":"openstack/project-config","branch":"master","topic":"nodepool-venvs","hashtags":[],"change_id":"I22c549b5f9b9e3882fcd2340946d2850b0b2f86b","subject":"nodepool elements: fix pip upgrade venv","status":"MERGED","created":"2022-07-06 03:32:09.000000000","updated":"2022-07-06 06:10:19.000000000","submitted":"2022-07-06 03:53:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"848800-nodepool-venvs","meta_rev_id":"793f537a8471bc5afc4624d333559d87ba178b3e","_number":848800,"virtual_id_number":848800,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-06 03:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-06 03:41:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-06 03:41:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-06 03:39:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d935daf5e123fbcb0ed9f89f0c55ce31a4c53448","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-06 03:32:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"640b029b7421dcfe221e748e93f8a1b92abb6975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 03:39:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7914bf55ad1d4f52a30254cfc5fc39b0\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6d20a11e02f34e8ba5e07345421830bf : SUCCESS in 2m 38s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bf870fc426cc4f1ca3f757b736bdcb3f : SUCCESS in 6m 27s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/6f7d55ab956640c89f8c886645f50d0b : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d1d400c139cdc3fb60c2b73e3144a78e00bed493","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-06 03:41:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"966c2dd393655831c15bdd86066c61ee8a78dcea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 03:42:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"279055ccb067727a30b91459dc9a3432cc0fc7d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 03:53:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39a54274ce844c93bf4bf59750f16c34\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/95510f65e496425b82e4df510dc86d01 : SUCCESS in 7m 13s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/e4f1d2b81c50412aae3ce949dda32aed : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"3e81ecb94a94f892e0b3603914d47bf8455892a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 03:53:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"793f537a8471bc5afc4624d333559d87ba178b3e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 06:10:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4269f780237e4dcd8259848cff152ac3\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/450a39197a8f464a9de46e9dcbd10dfa : SUCCESS in 9m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efbb9b896174f49d6dd35209153b66eaf2ec0d80","revisions":{"efbb9b896174f49d6dd35209153b66eaf2ec0d80":{"kind":"REWORK","_number":1,"created":"2022-07-06 03:32:09.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/00/848800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/00/848800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/848800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/848800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/00/848800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/00/848800/1"}}},"commit":{"parents":[{"commit":"e0143fb660641c6fef49eacbd6a3b4d19e97adcb","subject":"Merge \"Update openstack-ansible grafana dashboard job status rates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0143fb660641c6fef49eacbd6a3b4d19e97adcb"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-06 03:27:58.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-06 03:27:58.000000000","tz":600},"subject":"nodepool elements: fix pip upgrade venv","message":"nodepool elements: fix pip upgrade venv\n\nChange I316e9587b6e290cd421b47f506c91dbebe0975c0 had a rather\nembarrasing oversight in that it copied the /usr/bindep-env/bin/pip\ninvocation for upgrading pip to all the other venv\u0027s.\n\ni.e. we were upgrading hte bindep-env pip over and over, and not\nactually the pip in the working venv.  The os-testr install on older\nplatforms has now broken because it still tries to install with the\nancient inbuilt pip -- local testing has confirmed that it works with\nthe updated pip.\n\nChange-Id: I22c549b5f9b9e3882fcd2340946d2850b0b2f86b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efbb9b896174f49d6dd35209153b66eaf2ec0d80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/efbb9b896174f49d6dd35209153b66eaf2ec0d80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
