)]}'
{"id":"openstack%2Fproject-config~849120","triplet_id":"openstack%2Fproject-config~master~Id654f7a08f1e0ca0b432684d881bed53e4bb03d1","project":"openstack/project-config","branch":"master","topic":"ansible-5","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-07-11 02:23:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id654f7a08f1e0ca0b432684d881bed53e4bb03d1","subject":"Switch default Ansible version to 5 in all tenants","status":"MERGED","created":"2022-07-08 14:42:44.000000000","updated":"2022-07-11 03:51:55.000000000","submitted":"2022-07-11 02:23:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849120-ansible-5","meta_rev_id":"f07f7b35e9383fc6cda607fd7634df6e3c237f5e","_number":849120,"virtual_id_number":849120,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-11 02:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2022-07-08 14:50:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-10 22:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-11 02:12:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-08 14:50:54.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2022-07-08 14:51:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-10 22:44:17.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"c5a10bfe107476d02d490ac2a1306d7b0a829dc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-08 14:42:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e736333aaecad7e5838c729e12fe7739ddd152e","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2022-07-08 14:50:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e890fde6e13b760379536ddfd65d0b8ef73fe11c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 14:51:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16c2d5d8bf9e42fca0de420b71c6835f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/89800620037c4d5ca71ba6b9b7e13c85 : SUCCESS in 3m 16s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2b63342299be4f5d9f711265c5fa38aa : SUCCESS in 7m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5b8fc16bd8c94976bcf0a6a08533fc72 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8f894fe31d60a76c012d4cd9bc2e6517beeb2ab3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-10 22:44:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe2f0cde2a3bfb626ee78bdf046dc69497b7ca7a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-11 02:12:43.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e582662d3f01baf9e86470ac0ff5ed59c1939e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 02:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3a3e7c52518a91f1e0f55e4ed948a22e82e988d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 02:23:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c4e016f1a044c709dd75ec310cb95f1\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7a778a5a044a4f1690c48b06563ba017 : SUCCESS in 8m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/516a3ccd183a4825a11909156ff51b81 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"9d72aac3f240fc5c3d406bb99893ff1aa346e61b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 02:23:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f07f7b35e9383fc6cda607fd7634df6e3c237f5e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 03:51:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bc049fa793e4166809add447a27dfa6\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/ccebf60c74a44a14b7efd0c5a4cfd633 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94362e2aa695b2bb7d83cd52a9408381e489b0e6","revisions":{"94362e2aa695b2bb7d83cd52a9408381e489b0e6":{"kind":"REWORK","_number":1,"created":"2022-07-08 14:42:44.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/20/849120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/20/849120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/849120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/849120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/849120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/20/849120/1"}}},"commit":{"parents":[{"commit":"32d1b8f3004e8d0f30bbdcf4348250565791fc5b","subject":"Add ACL to allow collaboration with Trilio devs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/32d1b8f3004e8d0f30bbdcf4348250565791fc5b"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-07-08 14:40:55.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-07-08 14:40:55.000000000","tz":0},"subject":"Switch default Ansible version to 5 in all tenants","message":"Switch default Ansible version to 5 in all tenants\n\nIn our earlier announcement we said we\u0027d do this on June 30, which\nwas over a week ago. We should probably go ahead and get it done,\nbut maybe not over a weekend.\n\nhttps://lists.opendev.org/pipermail/service-announce/2022-June/000041.html\n\nChange-Id: Id654f7a08f1e0ca0b432684d881bed53e4bb03d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94362e2aa695b2bb7d83cd52a9408381e489b0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94362e2aa695b2bb7d83cd52a9408381e489b0e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
