)]}'
{"id":"openstack%2Fproject-config~850953","triplet_id":"openstack%2Fproject-config~master~Icf762f6aae68abf4e4885f7d48483d1698570706","project":"openstack/project-config","branch":"master","topic":"tobiko-dashboard","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-08-03 04:09:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf762f6aae68abf4e4885f7d48483d1698570706","subject":"Add Grafana dashboard for the Tobiko CI","status":"MERGED","created":"2022-07-25 21:57:14.000000000","updated":"2022-08-03 04:19:00.000000000","submitted":"2022-08-03 04:09:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850953-tobiko-dashboard","meta_rev_id":"e828e942a7f50b27aa9616f82131bfcae13eadd4","_number":850953,"virtual_id_number":850953,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16914,"name":"Federico Ressi","email":"federico.ressi@gmail.com","username":"serenico"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-03 04:09:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"recommended":{"_account_id":16914,"name":"Federico Ressi","email":"federico.ressi@gmail.com","username":"serenico"},"all":[{"value":1,"date":"2022-07-27 09:32:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16914,"name":"Federico Ressi","email":"federico.ressi@gmail.com","username":"serenico"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-27 04:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":16914,"name":"Federico Ressi","email":"federico.ressi@gmail.com","username":"serenico"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-03 04:01:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":16914,"name":"Federico Ressi","email":"federico.ressi@gmail.com","username":"serenico"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 22:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-27 04:51:57.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2022-07-27 09:32:17.000000000","updated_by":{"_account_id":16914,"name":"Federico Ressi","email":"federico.ressi@gmail.com","username":"serenico"},"reviewer":{"_account_id":16914,"name":"Federico Ressi","email":"federico.ressi@gmail.com","username":"serenico"},"state":"REVIEWER"}],"messages":[{"id":"4bb9ff66a285d4f9ebd732bb39c8cbee28d302c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-25 21:57:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c15080c55b305db1ddc8693291808094f7130ecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 22:21:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc5e46a9f474c3dbf6628c72a6b6ad0\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/63cf3b7d8b8d4b30a5348ce22200a071 : SUCCESS in 19m 27s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/47b2ae3aabc84cb394ee8af5547237a0 : SUCCESS in 7m 59s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/df660d659db44edf84b4ea41ad21b33d : SUCCESS in 3m 28s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/efb075bf9dd9461aa3507684a276cdce : SUCCESS in 12m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ca0d591fb83aa1eb2f2ac38d82fdbe41bb7c2592","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-27 04:51:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8e575edd66e5883875a695c88b6036cc2b5c3f1","author":{"_account_id":16914,"name":"Federico Ressi","email":"federico.ressi@gmail.com","username":"serenico"},"date":"2022-07-27 09:32:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd4601a9a364a914dedaa22a4ed413166a98580d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-03 04:01:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4837673b8f040fe7bc04c074bd8811e822bc4344","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 04:01:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87f361b215052fd5a7d7e4d606b96cbb0839ec13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 04:09:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aca4b07a618d47f48a2e902ac83f76d6\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3f0c870cc1fc43a99ebbdfe47ab50069 : SUCCESS in 7m 19s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/601169d5a15b4de78731b7dc14da4123 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"722568351ac86d48eb327e93e60f102bf980b1ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 04:09:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e828e942a7f50b27aa9616f82131bfcae13eadd4","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 04:19:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2067057689fd4831adb7f963c5e1a39d\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/afd21b78705c44fdb6c476de2ca40494 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6452e53157c5b59fd8e3c8afaa98bc50a3c2382f","revisions":{"6452e53157c5b59fd8e3c8afaa98bc50a3c2382f":{"kind":"REWORK","_number":1,"created":"2022-07-25 21:57:14.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/53/850953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/53/850953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/850953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/850953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/53/850953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/53/850953/1"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-25 21:56:16.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-25 21:57:06.000000000","tz":120},"subject":"Add Grafana dashboard for the Tobiko CI","message":"Add Grafana dashboard for the Tobiko CI\n\nChange-Id: Icf762f6aae68abf4e4885f7d48483d1698570706\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6452e53157c5b59fd8e3c8afaa98bc50a3c2382f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6452e53157c5b59fd8e3c8afaa98bc50a3c2382f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
