)]}'
{"id":"openstack%2Fproject-config~850962","triplet_id":"openstack%2Fproject-config~master~I547e5eb7faabcd9d3983ae85f3291b1a740bc77c","project":"openstack/project-config","branch":"master","topic":"proposal-remove-pip","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-07-26 21:56:43.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-07-26 23:07:55.000000000","reason":"Change was submitted"},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-07-26 23:07:55.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-07-26 19:07:32.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I547e5eb7faabcd9d3983ae85f3291b1a740bc77c","subject":"translation updates: refactor Python 3.8 workarounds","status":"MERGED","created":"2022-07-26 01:08:51.000000000","updated":"2022-07-27 15:12:25.000000000","submitted":"2022-07-26 23:07:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850962-proposal-remove-pip","meta_rev_id":"73ebe871b437f4e388bc693eb8892f36d337a0aa","_number":850962,"virtual_id_number":850962,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-26 23:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-07-26 21:56:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-26 23:00:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-26 23:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-26 01:16:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 15:30:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-07-26 15:30:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-07-26 15:30:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-07-26 15:30:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-07-26 20:15:24.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-07-27 15:12:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"b7ac614dd86674749fa5e289762709c25fde816a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-26 01:08:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e0e70967305115f97368039e1bdeb36625fdb78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 01:16:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/64ffc69ab6774cb5b823a282b12bfe72\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/413f695727204af199d623e219db4afb : SUCCESS in 3m 42s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/80f399e090a848f88fc9cfb1e9ee5cbc : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5673ef467d236b1af621e286373b7582df80c44d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-26 01:22:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee6601cd8fbf5c66c0d73557fa8069eca2cb761b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 01:38:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9205515acc04ae3a348e6cc23f4bcaa\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0312ea1d2cec4ab7bba8ccd67ea2e6a7 : SUCCESS in 7m 05s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d43d1e5a9a8149f0aca4dd1bf0ac1c80 : FAILURE in 8m 56s","accounts_in_message":[],"_revision_number":2},{"id":"3e28d32daf036dece1fa476fbb324eddf2a96a0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-26 03:03:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1cfbf95257889d03a58477fa23a863e113445adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 03:13:27.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"372a924aee22c8372b86c43a5727509599015c2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-26 04:55:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fa71900fb164983163cfb75f6393e3e1c614fffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 05:03:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d54395700d964a9283c5de1eb4505c90\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0dfb6c6c913c4a0ebaab5018b74d76a9 : SUCCESS in 7m 01s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a63f5142ab134584b992e3a14a0a83f3 : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":4},{"id":"f0f004ce9251387c5497e32e10024584fefdf2d7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-26 15:30:26.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"230d52641ed304cd56086c7c552492b318a4cf2a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-26 19:07:32.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ca3606dfa19054ff204dca4b70c35493ef53e879","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-07-26 20:15:24.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"82238bdfea096692062307b932bd16f29c7e145a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-07-26 21:19:03.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0912e3bb4251771ce4107c84a611d159e15b2478","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-26 21:41:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ac6e1d764247c401558e3dc3d9f346ee1b2a693c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-26 21:54:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"89081325eb33ae0a0a340fed1d0b20479b1d61bd","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-07-26 21:56:43.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"324dac01b3e1249da4c9d82a970ec1b404598740","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 22:02:52.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/613cd970ee3943e0831c3b8d9f867edc\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b615af8eed8a47de8c3fa84af1379da8 : SUCCESS in 3m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2649dc133cf04f8395a48b58da0ff625 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":5},{"id":"67d73d52f4b24a821056d2ca3e3dfae99a8e39f3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-26 23:00:07.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"86965cd215255086e99ae139e412012325f4caf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 23:00:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"86f9c70597e7c63be8f624579bf0b61dc1f10459","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 23:07:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cc71d15909f41b68eef7337c1d70412\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/071012d2c9a64938978b935969c8d9f8 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":5},{"id":"488d462478df83356c1cb59e34fb3b6c3b8cf22b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 23:07:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"73ebe871b437f4e388bc693eb8892f36d337a0aa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-27 15:12:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6f1e9fb2b9f9f6edaa6111120fbf9f12a534f225","revisions":{"837ed81afd97f24f066803b1ff5eee734a47ff69":{"kind":"REWORK","_number":1,"created":"2022-07-26 01:08:51.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/62/850962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/850962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/850962/1"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 00:47:44.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 01:05:39.000000000","tz":600},"subject":"translation updates: refactor Python 3.8 workarounds","message":"translation updates: refactor Python 3.8 workarounds\n\nI think we have quite a tangle of fixes on fixes in the translation\njobs, and I hope this refactor makes things a bit better.\n\nThe fundamental problem seems to be that python3.6 can\u0027t work with\nmaster requirements.txt.  But we can not move some jobs off bionic\n(python 3.6) because of java version contsraints.\n\nThe first thing that happened was these jobs got \"ensure-python\" with\n\"python_version: 3.8\" set to bring in python3.8 on Bionic.  This role\nwas not bringing in python3.8-venv, so we had no way to create 3.8\nvirtual environments.  Thus ensure-pip ended up getting modified to\ndrag this package in (reverted with notes in\n372f3af706c796331578ef81c2680e8c4c03c3b6).  This has a fairly easy\nsolution; let\u0027s just make ensure-python bring in the venv packages for\nthe udpated python -- see dependency\nIe3c03fea82bcec80a897f0905c15f35405a50396.\n\nThe git-review install here uses \u0027ensure_pip_virtualenv_command\u0027 to\ninstall git-review in a virtualenv.  git-review still supports bionic,\nand does not need to be installed in any speical way.  However, by\noverriding the \u0027ensure_pip_virtualenv_command\u0027 for the entire job,\nthis unrelated install now grew a dependency on python3.8 and created\nmore confusion over why it was failing.\n\nI believe what we actually want to do is just run tox under python3.8.\nSo this pulls out all the job-level defintions and encapsulates them\nin a single block scoped to run on bionic.  It will pull in python3.8\npackages, and then install tox under python3.8 with an override of the\nvirtulenv command just for that step.  The jobs just maintain their\nnode definition and note on why it is required.\n\nThe long-term future of these jobs is obviously a question; but I\nbelieve this keeps them running on bionic in as sane a way as possible\nfor now.\n\nDepends-On: Ie3c03fea82bcec80a897f0905c15f35405a50396\nChange-Id: I547e5eb7faabcd9d3983ae85f3291b1a740bc77c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/837ed81afd97f24f066803b1ff5eee734a47ff69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/837ed81afd97f24f066803b1ff5eee734a47ff69"}]},"branch":"refs/heads/master"},"84853c495463728f07c6c5100b07457f2d97d187":{"kind":"REWORK","_number":2,"created":"2022-07-26 01:22:01.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/62/850962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/850962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/850962/2"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 00:47:44.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 01:20:59.000000000","tz":600},"subject":"translation updates: refactor Python 3.8 workarounds","message":"translation updates: refactor Python 3.8 workarounds\n\nI think we have quite a tangle of fixes on fixes in the translation\njobs, and I hope this refactor makes things a bit better.\n\nThe fundamental problem seems to be that python3.6 can\u0027t work with\nmaster requirements.txt.  But we can not move some jobs off bionic\n(python 3.6) because of java version contsraints.\n\nThe first thing that happened was these jobs got \"ensure-python\" with\n\"python_version: 3.8\" set to bring in python3.8 on Bionic.  This role\nwas not bringing in python3.8-venv, so we had no way to create 3.8\nvirtual environments.  Thus ensure-pip ended up getting modified to\ndrag this package in (reverted with notes in\n372f3af706c796331578ef81c2680e8c4c03c3b6).  This has a fairly easy\nsolution; let\u0027s just make ensure-python bring in the venv packages for\nthe udpated python -- see dependency\nIe3c03fea82bcec80a897f0905c15f35405a50396.\n\nThe git-review install here uses \u0027ensure_pip_virtualenv_command\u0027 to\ninstall git-review in a virtualenv.  git-review still supports bionic,\nand does not need to be installed in any speical way.  However, by\noverriding the \u0027ensure_pip_virtualenv_command\u0027 for the entire job,\nthis unrelated install now grew a dependency on python3.8 and created\nmore confusion over why it was failing.\n\nI believe what we actually want to do is just run tox under python3.8.\nSo this pulls out all the job-level defintions and encapsulates them\nin a single block scoped to run on bionic.  It will pull in python3.8\npackages, and then install tox under python3.8 with an override of the\nvirtulenv command just for that step.  The jobs just maintain their\nnode definition and note on why it is required.\n\nThe long-term future of these jobs is obviously a question; but I\nbelieve this keeps them running on bionic in as sane a way as possible\nfor now.\n\nDepends-On: Ie3c03fea82bcec80a897f0905c15f35405a50396\nChange-Id: I547e5eb7faabcd9d3983ae85f3291b1a740bc77c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84853c495463728f07c6c5100b07457f2d97d187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84853c495463728f07c6c5100b07457f2d97d187"}]},"branch":"refs/heads/master"},"78257cd0dad3aaa8716c67075a4a6ed389477a29":{"kind":"REWORK","_number":3,"created":"2022-07-26 03:03:06.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/62/850962/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/850962/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/850962/3"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 00:47:44.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 03:02:53.000000000","tz":600},"subject":"translation updates: refactor Python 3.8 workarounds","message":"translation updates: refactor Python 3.8 workarounds\n\nI think we have quite a tangle of fixes on fixes in the translation\njobs, and I hope this refactor makes things a bit better.\n\nThe fundamental problem seems to be that python3.6 can\u0027t work with\nmaster requirements.txt.  But we can not move some jobs off bionic\n(python 3.6) because of java version contsraints.\n\nThe first thing that happened was these jobs got \"ensure-python\" with\n\"python_version: 3.8\" set to bring in python3.8 on Bionic.  This role\nwas not bringing in python3.8-venv, so we had no way to create 3.8\nvirtual environments.  Thus ensure-pip ended up getting modified to\ndrag this package in (reverted with notes in\n372f3af706c796331578ef81c2680e8c4c03c3b6).  This has a fairly easy\nsolution; let\u0027s just make ensure-python bring in the venv packages for\nthe udpated python -- see dependency\nIe3c03fea82bcec80a897f0905c15f35405a50396.\n\nThe git-review install here uses \u0027ensure_pip_virtualenv_command\u0027 to\ninstall git-review in a virtualenv.  git-review still supports bionic,\nand does not need to be installed in any speical way.  However, by\noverriding the \u0027ensure_pip_virtualenv_command\u0027 for the entire job,\nthis unrelated install now grew a dependency on python3.8 and created\nmore confusion over why it was failing.\n\nI believe what we actually want to do is just run tox under python3.8.\nSo this pulls out all the job-level defintions and encapsulates them\nin a single block scoped to run on bionic.  It will pull in python3.8\npackages, and then install tox under python3.8 with an override of the\nvirtulenv command just for that step.  The jobs just maintain their\nnode definition and note on why it is required.\n\nThe long-term future of these jobs is obviously a question; but I\nbelieve this keeps them running on bionic in as sane a way as possible\nfor now.\n\nDepends-On: Ie3c03fea82bcec80a897f0905c15f35405a50396\nChange-Id: I547e5eb7faabcd9d3983ae85f3291b1a740bc77c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/78257cd0dad3aaa8716c67075a4a6ed389477a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/78257cd0dad3aaa8716c67075a4a6ed389477a29"}]},"branch":"refs/heads/master"},"2139e943daea700a1210e8d85838a88a13aa4209":{"kind":"REWORK","_number":4,"created":"2022-07-26 04:55:12.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/62/850962/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/850962/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/850962/4"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 00:47:44.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 04:55:01.000000000","tz":600},"subject":"translation updates: refactor Python 3.8 workarounds","message":"translation updates: refactor Python 3.8 workarounds\n\nI think we have quite a tangle of fixes on fixes in the translation\njobs, and I hope this refactor makes things a bit better.\n\nThe fundamental problem seems to be that python3.6 can\u0027t work with\nmaster requirements.txt.  But we can not move some jobs off bionic\n(python 3.6) because of java version contsraints.\n\nThe first thing that happened was these jobs got \"ensure-python\" with\n\"python_version: 3.8\" set to bring in python3.8 on Bionic.  This role\nwas not bringing in python3.8-venv, so we had no way to create 3.8\nvirtual environments.  Thus ensure-pip ended up getting modified to\ndrag this package in (reverted with notes in\n372f3af706c796331578ef81c2680e8c4c03c3b6).  This has a fairly easy\nsolution; let\u0027s just make ensure-python bring in the venv packages for\nthe udpated python -- see dependency\nIe3c03fea82bcec80a897f0905c15f35405a50396.\n\nThe git-review install here uses \u0027ensure_pip_virtualenv_command\u0027 to\ninstall git-review in a virtualenv.  git-review still supports bionic,\nand does not need to be installed in any speical way.  However, by\noverriding the \u0027ensure_pip_virtualenv_command\u0027 for the entire job,\nthis unrelated install now grew a dependency on python3.8 and created\nmore confusion over why it was failing.\n\nI believe what we actually want to do is just run tox under python3.8.\nSo this pulls out all the job-level defintions and encapsulates them\nin a single block scoped to run on bionic.  It will pull in python3.8\npackages, and then install tox under python3.8 with an override of the\nvirtulenv command just for that step.  The jobs just maintain their\nnode definition and note on why it is required.\n\nThe long-term future of these jobs is obviously a question; but I\nbelieve this keeps them running on bionic in as sane a way as possible\nfor now.\n\nDepends-On: Ie3c03fea82bcec80a897f0905c15f35405a50396\nChange-Id: I547e5eb7faabcd9d3983ae85f3291b1a740bc77c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2139e943daea700a1210e8d85838a88a13aa4209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2139e943daea700a1210e8d85838a88a13aa4209"}]},"branch":"refs/heads/master"},"6f1e9fb2b9f9f6edaa6111120fbf9f12a534f225":{"kind":"REWORK","_number":5,"created":"2022-07-26 21:54:38.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/62/850962/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/850962/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/850962/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/850962/5"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 00:47:44.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 21:48:43.000000000","tz":600},"subject":"translation updates: refactor Python 3.8 workarounds","message":"translation updates: refactor Python 3.8 workarounds\n\nI think we have quite a tangle of fixes on fixes in the translation\njobs, and I hope this refactor makes things a bit better.\n\nThe fundamental problem seems to be that python3.6 can\u0027t work with\nmaster requirements.txt.  But we can not move some jobs off bionic\n(python 3.6) because of java version contsraints.\n\nThe first thing that happened was these jobs got \"ensure-python\" with\n\"python_version: 3.8\" set to bring in python3.8 on Bionic.  This role\nwas not bringing in python3.8-venv, so we had no way to create 3.8\nvirtual environments.  Thus ensure-pip ended up getting modified to\ndrag this package in (reverted with notes in\n372f3af706c796331578ef81c2680e8c4c03c3b6).  This has a fairly easy\nsolution; let\u0027s just make ensure-python bring in the venv packages for\nthe udpated python -- see dependency\nIe3c03fea82bcec80a897f0905c15f35405a50396.\n\nThe git-review install here uses \u0027ensure_pip_virtualenv_command\u0027 to\ninstall git-review in a virtualenv.  git-review still supports bionic,\nand does not need to be installed in any speical way.  However, by\noverriding the \u0027ensure_pip_virtualenv_command\u0027 for the entire job,\nthis unrelated install now grew a dependency on python3.8 and created\nmore confusion over why it was failing.\n\nI believe what we actually want to do on Bionic is just run tox under\npython3.8.  So this pulls out all the job-level defintions and\nencapsulates them in a single block scoped to run on bionic.  It will\npull in python3.8 packages, and then install tox under python3.8 with\nan override of the virtulenv command just for that step.  The jobs\njust maintain their node definition and note on why it is required.\n\nThe long-term future of these jobs is obviously a question; but I\nbelieve this keeps them running on bionic in as sane a way as possible\nfor now.\n\nThe second part of this is the python3.9 interpreter required on\nFocal.  Setting \"python_version: 3.9\" in the job definition is a bit\nconfusing, because the job actually requires both Python 3.8 AND 3.9.\nThis variable is passed to ensure-python, and the naming is a bit\nunfortunate because it is so broad (but we can\u0027t change that without\nzuul-jobs updates).  Here I\u0027ve pulled out the Python 3.9 install into\na separate focal-gated section in the pre.yaml that explains what is\ngoing on (note that Python 3.8 is just the base python on focal, so\nneeds no extra installation).  I\u0027ve also updated the description on\nthe job node with this info.  There\u0027s no need to install the tox\nenvironment any differently on focal; it just uses the base system\npython (3.8) which is fine.\n\nDepends-On: https://review.opendev.org/c/zuul/zuul-jobs/+/850957\nChange-Id: I547e5eb7faabcd9d3983ae85f3291b1a740bc77c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f1e9fb2b9f9f6edaa6111120fbf9f12a534f225"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f1e9fb2b9f9f6edaa6111120fbf9f12a534f225"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
