)]}'
{"id":"openstack%2Fproject-config~851031","triplet_id":"openstack%2Fproject-config~master~Ie0c7bc4558e4e73910664b6796d5ada7848bf5d2","project":"openstack/project-config","branch":"master","topic":"tripleo-neutron-tempest-plugin","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-08-08 11:57:58.000000000","reason":"Change was submitted"},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-08-08 11:57:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0c7bc4558e4e73910664b6796d5ada7848bf5d2","subject":"Add new project \"whitebox-neutron-tempest-plugin\" in the x/ namespace","status":"MERGED","created":"2022-07-26 13:39:10.000000000","updated":"2022-08-08 12:16:34.000000000","submitted":"2022-08-08 11:57:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851031-tripleo-neutron-tempest-plugin","meta_rev_id":"d1f88941482d2dbddb9da039a375a2fc55ed8409","_number":851031,"virtual_id_number":851031,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 11:57:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"all":[{"value":1,"date":"2022-07-29 13:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-08-01 07:37:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-29 09:37:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"value":1,"date":"2022-08-01 07:52:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":2,"date":"2022-08-02 22:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},{"value":0,"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"value":1,"date":"2022-08-08 11:49:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-26 13:48:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-26 14:53:27.000000000","updated_by":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"reviewer":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"state":"REVIEWER"},{"updated":"2022-07-26 15:20:13.000000000","updated_by":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"reviewer":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"state":"REVIEWER"},{"updated":"2022-07-29 13:35:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-08-01 07:52:14.000000000","updated_by":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"reviewer":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"state":"REVIEWER"},{"updated":"2022-08-02 22:15:09.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-08-08 07:22:03.000000000","updated_by":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"8b81ed4ec8430c4bbebf52d49f552337e78a8fbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-26 13:39:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86ba840c99482f93767862f8fc63010b83d20082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 13:48:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b0e76dbf1e3401fbfbeaabff1bce34c\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1a313827e5c24d8d8dae09dd876c2c4e : SUCCESS in 4m 16s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/73cdbb6f7a9549c9bc8a8d7cc54f0e67 : SUCCESS in 7m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4c3bdce891e14f27b72c970d7ccea5af : FAILURE in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1cf41e9f9cbaa04d5d221e4c1eac2e7eb1ce0929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-26 14:07:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"684f2a6eb5579c5e52d2b90b72eef4240e012b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 14:16:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a56743edc93842b2b5ed1a80cfa4f3d6\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6170e954c7fb459295bf015dbb8bceb5 : SUCCESS in 4m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3494fb50365644c99e4a657d0e6dbe11 : SUCCESS in 7m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d8b799151da7438cbc2467cc06b50e53 : FAILURE in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ba2c542e7920b6f3b4448243b5981395a477aef2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-26 14:37:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c8e39cd4e3f2d04cdbf104696b43edee4eeae0a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-26 14:43:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"db7fbe18bb5a34603fc29a9647816a7f0d82803e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 14:51:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e0d09a4c764c98a0eac87b67a8d5fe\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1bf2bef5710946ed85d33b84f209ab51 : SUCCESS in 2m 45s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/312c827f36f14178bac48d013302416b : SUCCESS in 7m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fe56618a075c4329943f68b3e7b5b4d3 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":4},{"id":"75e986f77def919c060d7d5eda7c0e6aa4bf0ae4","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2022-07-26 14:53:27.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"60d56b193d4e2315f37b1a124952353e861a82fd","author":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"date":"2022-07-26 15:20:13.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"f2fc6214b9011273b07cd5bdc7b3abb4ebca32ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-29 07:41:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b688307cbe76b4ac1a37eab0e55ab9f2ccf1aca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 07:49:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c02884f71934938ac30422923dca81e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/74f138cc714c40d1bd0c5a3b94da492d : SUCCESS in 4m 32s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/44c5f324f9744bff90cd9e0b9ef8b3d8 : SUCCESS in 7m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f2aa3fa12e444059b459896d4585208d : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":5},{"id":"9cf4b81ecc49fd29c64301c0eb777b400e41a7c8","author":{"_account_id":33341,"name":"Maor Blaustein","email":"mblue@redhat.com","username":"blue"},"date":"2022-07-29 09:37:46.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"224bc21b98ed711b307b9d1b44a098b7b148d5df","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-29 13:35:06.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3449669fc51be20ce3936a96c75b8a563abd6049","author":{"_account_id":31450,"name":"Alex Katz","email":"akatz@redhat.com","username":"akatz"},"date":"2022-08-01 07:37:44.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c9d6ad10f4010adbf02c9e6c29ba05605ccee7df","author":{"_account_id":19118,"name":"Eran Kuris","email":"ekuris@redhat.com","username":"ekuris"},"date":"2022-08-01 07:52:14.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bbdc580752d1b72b1b86df1d46da609c4cbc7556","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-02 22:15:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"fa285530eeaaf6bd9ef748aa7bae345cd97bdd24","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-08 11:49:10.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5de0c0ada03fbc5c5b6fa9737e16afa9591e6a24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 11:49:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c92c3c821e0e33da5de63743ac880bc1575ab18d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 11:57:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1077c1b425f4ec48fdad2e0f43b7885\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1a9d6abe2c87401f88c7cb00c3862d15 : SUCCESS in 7m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9f8fc4164d3e4efab5e1f2c95e65a459 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":5},{"id":"fba8e4a2ba71c1c66c32ffc77de90cc92d1f0c1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 11:57:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"d1f88941482d2dbddb9da039a375a2fc55ed8409","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 12:16:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/100f86be288741cb975c60046670cd37\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/c7503b2e76454379aadab655439b0ffa : SUCCESS in 8m 06s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/4d1e1c5e31ec4562bc63ea810e848b98 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bf358222715cf1bde83c8446b3ec8869826e07f2","revisions":{"075da62a8dad6a348b205086bbcf2d662bfdf0e0":{"kind":"REWORK","_number":1,"created":"2022-07-26 13:39:10.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/31/851031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/851031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/851031/1"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-26 13:38:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-26 13:38:02.000000000","tz":120},"subject":"Add new project \"tripleo-neutron-tempest-plugin\" in the x/ namespace","message":"Add new project \"tripleo-neutron-tempest-plugin\" in the x/ namespace\n\nThis project will contain neutron related tempest tests useful for the\nTripleo based deployments, which don\u0027t fit well into official\nneutron-tempest-plugin repository.\n\nChange-Id: Ie0c7bc4558e4e73910664b6796d5ada7848bf5d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/075da62a8dad6a348b205086bbcf2d662bfdf0e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/075da62a8dad6a348b205086bbcf2d662bfdf0e0"}]},"branch":"refs/heads/master"},"1604734e276501a0537b2f756920085ae674a8b0":{"kind":"REWORK","_number":2,"created":"2022-07-26 14:07:42.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/31/851031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/851031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/851031/2"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-26 13:38:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-26 14:07:33.000000000","tz":120},"subject":"Add new project \"tripleo-neutron-tempest-plugin\" in the x/ namespace","message":"Add new project \"tripleo-neutron-tempest-plugin\" in the x/ namespace\n\nThis project will contain neutron related tempest tests useful for the\nTripleo based deployments, which don\u0027t fit well into official\nneutron-tempest-plugin repository.\n\nChange-Id: Ie0c7bc4558e4e73910664b6796d5ada7848bf5d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1604734e276501a0537b2f756920085ae674a8b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1604734e276501a0537b2f756920085ae674a8b0"}]},"branch":"refs/heads/master"},"c3073be08a6d799de6b22179b636ad1d69803079":{"kind":"REWORK","_number":3,"created":"2022-07-26 14:37:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/31/851031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/851031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/851031/3"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-26 13:38:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-26 14:37:02.000000000","tz":120},"subject":"Add new project \"tripleo-neutron-tempest-plugin\" in the x/ namespace","message":"Add new project \"tripleo-neutron-tempest-plugin\" in the x/ namespace\n\nThis project will contain neutron related tempest tests useful for the\nTripleo based deployments, which don\u0027t fit well into official\nneutron-tempest-plugin repository.\n\nChange-Id: Ie0c7bc4558e4e73910664b6796d5ada7848bf5d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c3073be08a6d799de6b22179b636ad1d69803079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c3073be08a6d799de6b22179b636ad1d69803079"}]},"branch":"refs/heads/master"},"a34d983731a8272efd4e05e78fa6c2c1f4fe0707":{"kind":"REWORK","_number":4,"created":"2022-07-26 14:43:11.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/31/851031/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/851031/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/851031/4"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-26 13:38:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-26 14:43:03.000000000","tz":120},"subject":"Add new project \"tripleo-neutron-tempest-plugin\" in the x/ namespace","message":"Add new project \"tripleo-neutron-tempest-plugin\" in the x/ namespace\n\nThis project will contain neutron related tempest tests useful for the\nTripleo based deployments, which don\u0027t fit well into official\nneutron-tempest-plugin repository.\n\nChange-Id: Ie0c7bc4558e4e73910664b6796d5ada7848bf5d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a34d983731a8272efd4e05e78fa6c2c1f4fe0707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a34d983731a8272efd4e05e78fa6c2c1f4fe0707"}]},"branch":"refs/heads/master"},"bf358222715cf1bde83c8446b3ec8869826e07f2":{"kind":"REWORK","_number":5,"created":"2022-07-29 07:41:52.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/31/851031/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/851031/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/851031/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/851031/5"}}},"commit":{"parents":[{"commit":"b863397eed547dfc0bb5f86739fee853cb3fa98c","subject":"VPNAAS: remove old jobs from grafana","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b863397eed547dfc0bb5f86739fee853cb3fa98c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-26 13:38:02.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-29 07:41:21.000000000","tz":120},"subject":"Add new project \"whitebox-neutron-tempest-plugin\" in the x/ namespace","message":"Add new project \"whitebox-neutron-tempest-plugin\" in the x/ namespace\n\nThis project will contain whitebox neutron related tempest tests useful\nfor the Tripleo based deployments, which don\u0027t fit well into official\nneutron-tempest-plugin repository.\n\nChange-Id: Ie0c7bc4558e4e73910664b6796d5ada7848bf5d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf358222715cf1bde83c8446b3ec8869826e07f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf358222715cf1bde83c8446b3ec8869826e07f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
