)]}'
{"id":"openstack%2Fproject-config~851098","triplet_id":"openstack%2Fproject-config~master~I24b9e5628823dce83310f96a8e474a0f0756f343","project":"openstack/project-config","branch":"master","topic":"proposal-remove-pip","hashtags":[],"change_id":"I24b9e5628823dce83310f96a8e474a0f0756f343","subject":"[dnm] additional discussion for requirements proposal","status":"ABANDONED","created":"2022-07-26 21:35:50.000000000","updated":"2022-07-26 21:59:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7f36d52f37575825163fd480071afdfb9895d0fa","_number":851098,"virtual_id_number":851098,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-26 21:44:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-26 21:44:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fdee37b237167d62c69cea9d499cee90214f5c30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-26 21:35:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6d81f36b6ae75c2bb1c7fa0cc5ebd12395323dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 21:44:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086be92fdb1048e9948081c027af66d8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8dff05f3a3a248fdb8ac1fd5d3586f9e : SUCCESS in 7m 32s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9442921156374d75a262dac30a1eee66 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"7f36d52f37575825163fd480071afdfb9895d0fa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-26 21:59:17.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/project-config/+/850962","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c93e8a9f31c5c758cee92ba29db8aa2bf95d9c24","revisions":{"c93e8a9f31c5c758cee92ba29db8aa2bf95d9c24":{"kind":"REWORK","_number":1,"created":"2022-07-26 21:35:50.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/98/851098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/98/851098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/851098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/851098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/98/851098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/98/851098/1"}}},"commit":{"parents":[{"commit":"2139e943daea700a1210e8d85838a88a13aa4209","subject":"translation updates: refactor Python 3.8 workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2139e943daea700a1210e8d85838a88a13aa4209"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 21:35:18.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-26 21:35:18.000000000","tz":600},"subject":"[dnm] additional discussion for requirements proposal","message":"[dnm] additional discussion for requirements proposal\n\nChange-Id: I24b9e5628823dce83310f96a8e474a0f0756f343\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c93e8a9f31c5c758cee92ba29db8aa2bf95d9c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c93e8a9f31c5c758cee92ba29db8aa2bf95d9c24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
