)]}'
{"id":"openstack%2Fproject-config~851156","triplet_id":"openstack%2Fproject-config~master~I074f089371054856a8b96bd6e650496a0bd53685","project":"openstack/project-config","branch":"master","topic":"proposal-remove-pip","hashtags":[],"change_id":"I074f089371054856a8b96bd6e650496a0bd53685","subject":"proposal updates: fix typo in release match","status":"MERGED","created":"2022-07-27 04:50:02.000000000","updated":"2022-07-27 05:11:01.000000000","submitted":"2022-07-27 05:11:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851156-proposal-remove-pip","meta_rev_id":"fa141780ff02aad5fe00a366ad9e4fb73673322c","_number":851156,"virtual_id_number":851156,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-27 05:11:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-27 04:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-27 04:50:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-27 05:03:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8787d32ddf3fadfd9da9137b0c64bbdf75f250fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-27 04:50:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0cbaa609308791f15447c54454d557c60f2e8ac","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-27 04:50:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2a89c33f01fb34cfc953af532066541ce16b328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 05:03:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca538dacb3784b6f9e314e445102cded\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0a489ec6e2624bc1a1dd64d6a1011e3e : SUCCESS in 3m 42s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e7a502c3a344c32b6ebda7cbe3f1fc4 : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"45b303b443b90aa79342352b29fb578fbe7bd54e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 05:03:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2769c009d805e9850daf87a6080e26c76c291ccb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 05:11:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f97f1aa9c8d4c0f82ea01c231950228\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/860d4854286e4f56a9423f45885461af : SUCCESS in 7m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fa141780ff02aad5fe00a366ad9e4fb73673322c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 05:11:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4616695be07dd5e4c3ec4db716792493cdc46cec","revisions":{"4616695be07dd5e4c3ec4db716792493cdc46cec":{"kind":"REWORK","_number":1,"created":"2022-07-27 04:50:02.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/56/851156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/851156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/851156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/851156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/851156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/851156/1"}}},"commit":{"parents":[{"commit":"6f1e9fb2b9f9f6edaa6111120fbf9f12a534f225","subject":"translation updates: refactor Python 3.8 workarounds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f1e9fb2b9f9f6edaa6111120fbf9f12a534f225"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-27 04:49:09.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-27 04:49:09.000000000","tz":600},"subject":"proposal updates: fix typo in release match","message":"proposal updates: fix typo in release match\n\nThis fixes a typo in I683a61b1bae1a809caf724aee87a21af2e18fb0c -- it\nshould be \"ansible_distribution_release\"\n\nChange-Id: I074f089371054856a8b96bd6e650496a0bd53685\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4616695be07dd5e4c3ec4db716792493cdc46cec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4616695be07dd5e4c3ec4db716792493cdc46cec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
