)]}'
{"id":"openstack%2Fproject-config~851276","triplet_id":"openstack%2Fproject-config~master~Ia87a0c21ec0ed1662e37cbc9e17a0df344b54e57","project":"openstack/project-config","branch":"master","topic":"ansible-lint-update-6","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-08-04 11:53:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia87a0c21ec0ed1662e37cbc9e17a0df344b54e57","subject":"linter: add name to blocks","status":"MERGED","created":"2022-07-28 02:10:22.000000000","updated":"2022-08-04 11:53:57.000000000","submitted":"2022-08-04 11:53:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851276-ansible-lint-update-6","meta_rev_id":"221fce3e72f7b4cd4bec98f5f904ca67e4933903","_number":851276,"virtual_id_number":851276,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 11:53:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-08-02 21:29:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-04 11:41:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-04 11:41:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 02:46:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-02 21:29:12.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-08-04 11:41:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"7eeb5f207c6c66d9275e681b2d4db5b2c9509ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-28 02:10:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e97a5b6872839be7304d94cbe6c0aed028ca1c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 02:46:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77bce7e3e823426ea86c88d1ed650f97\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d03486b2f6cf408bb1575cb604cd0c0d : SUCCESS in 3m 16s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7f66a3c5612a48f8b4eeab842923dd14 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4534f832c8023a7a4e054d95a85d922e8af7ae96","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-28 03:57:29.000000000","message":"Topic ansible-lint-udpate removed","accounts_in_message":[],"_revision_number":1},{"id":"b89cd2db82f6a393cbe4aedab7082c090083fc33","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-07-28 03:57:35.000000000","message":"Topic set to ansible-lint-update-6","accounts_in_message":[],"_revision_number":1},{"id":"4155a27c32b0814609b7a168392dd727c60f116b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-08-02 21:29:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"73cba5f6fae6b0de3b2d0ce79d8fea30c7b6adc9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-04 11:41:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6ac32068619d846d70bb076824d053b20af6414","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 11:42:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0bbfe3f496f6981c0c062e2c06da460469ef48c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 11:53:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7391f54b4ef4d82a31f9a9823de8fbb\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8c8a55304b364c70abe231297fb02ac5 : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"221fce3e72f7b4cd4bec98f5f904ca67e4933903","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 11:53:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76037c434e65cd5eb207a281eb33bce6cf771c34","revisions":{"76037c434e65cd5eb207a281eb33bce6cf771c34":{"kind":"REWORK","_number":1,"created":"2022-07-28 02:10:22.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/76/851276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/76/851276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/851276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/851276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/76/851276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/76/851276/1"}}},"commit":{"parents":[{"commit":"8cdf719bb3f97779af56db4167718dde76d1d8e9","subject":"linter: update some YAML whitespace issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8cdf719bb3f97779af56db4167718dde76d1d8e9"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-28 01:14:11.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-07-28 02:03:23.000000000","tz":600},"subject":"linter: add name to blocks","message":"linter: add name to blocks\n\nA later version of ansible-lint enforces names on blocks.  This is\ngenerally a good rule; fix a few missing blocks here.\n\nChange-Id: Ia87a0c21ec0ed1662e37cbc9e17a0df344b54e57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76037c434e65cd5eb207a281eb33bce6cf771c34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/76037c434e65cd5eb207a281eb33bce6cf771c34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
