)]}'
{"id":"openstack%2Fproject-config~851519","triplet_id":"openstack%2Fproject-config~master~I42adc68695cd1fdf48779674935de69b81ffb9a2","project":"openstack/project-config","branch":"master","topic":"rockylinux-8","hashtags":[],"change_id":"I42adc68695cd1fdf48779674935de69b81ffb9a2","subject":"Temporarily turn on console logs for rocky in iweb","status":"MERGED","created":"2022-07-29 13:59:19.000000000","updated":"2022-07-29 15:02:46.000000000","submitted":"2022-07-29 14:31:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851519-rockylinux-8","meta_rev_id":"ced8c2fe7ab805b77f799e678e4083c75bde540e","_number":851519,"virtual_id_number":851519,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-29 14:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-29 14:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-29 14:18:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-29 14:11:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cfe829271759b861cb4fe909469aafe140c63f77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-29 13:59:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6813e03a499204436e728d93fcbc0e5d12a1a25e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:11:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72dbf35354df4dc0871ba4e81b52eeb5\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/77b09f88d91f43e4acc3198787a5ffbb : SUCCESS in 3m 08s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/10b274c92f44484dba11bb936e41b1d6 : SUCCESS in 7m 37s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/85a9713a004443eca7ae97ae10a0953b : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"83fd7141629e9f58ddbbade2803c28a8eb969962","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-07-29 14:18:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSelf-approving this so we can start collecting logs from it sooner.","accounts_in_message":[],"_revision_number":1},{"id":"b73e8de78979e0bf72743b4b375566ec2db5253d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:19:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92c6332a95f1086840c639e39a9b46b99a4b3b95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:31:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84948347d8464e9db7d1b91e93f52803\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1186b08f41884c589c790ea0ea824c5e : SUCCESS in 8m 32s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/84a225b60e474b0ea965d9a8005ab0c3 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1},{"id":"05af38b51156621371e595a48675229885348498","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:31:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ced8c2fe7ab805b77f799e678e4083c75bde540e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 15:02:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61c74f35e94841bd891d4082b74225de\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/1ee5a8b4fb6344448f412289d22b74dc : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1a58f0f325958c9525d4e6759118b7021b5b99b2","revisions":{"1a58f0f325958c9525d4e6759118b7021b5b99b2":{"kind":"REWORK","_number":1,"created":"2022-07-29 13:59:19.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/19/851519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/851519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/851519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/851519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/851519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/851519/1"}}},"commit":{"parents":[{"commit":"c1b692b0a0793d64207efc387e585ee65c763ebf","subject":"Synchronize diskimages on builders","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1b692b0a0793d64207efc387e585ee65c763ebf"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-07-29 13:56:02.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-07-29 13:56:02.000000000","tz":0},"subject":"Temporarily turn on console logs for rocky in iweb","message":"Temporarily turn on console logs for rocky in iweb\n\nOur rockylinux-8 images have stopped being reachable on boot over\nthe past few days. The images are building fine, but we need more\ninsight into what\u0027s happening at boot time. The only provider with\nstandard flavors, console log capability and sufficient quota to\nboot these with any regularity is iweb-mtl01, so turn on the\nconsole-log toggle there for now. We can revert this once the root\ncause is identified.\n\nChange-Id: I42adc68695cd1fdf48779674935de69b81ffb9a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a58f0f325958c9525d4e6759118b7021b5b99b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a58f0f325958c9525d4e6759118b7021b5b99b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
