)]}'
{"id":"openstack%2Fproject-config~851857","triplet_id":"openstack%2Fproject-config~master~I11fc532aab589d76525081e314fad989f244cc64","project":"openstack/project-config","branch":"master","topic":"osh-addons","attention_set":{},"removed_from_attention_set":{"8898":{"account":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"last_update":"2022-08-09 05:18:05.000000000","reason":"\u003cGERRIT_ACCOUNT_8898\u003e replied on the change","reason_account":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"}},"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-08-11 16:07:14.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-08-11 15:53:02.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I11fc532aab589d76525081e314fad989f244cc64","subject":"End project gating for openstack-helm-addons","status":"MERGED","created":"2022-08-02 13:37:17.000000000","updated":"2022-08-11 16:07:14.000000000","submitted":"2022-08-11 16:07:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851857-osh-addons","meta_rev_id":"e135426d7c297654073448ee9669b1c52ca6861c","_number":851857,"virtual_id_number":851857,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 16:07:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2022-08-09 05:18:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":1,"date":"2022-08-11 15:53:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-11 15:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-11 15:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 13:49:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 14:27:49.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"state":"REVIEWER"},{"updated":"2022-08-06 19:11:15.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-08-11 15:58:21.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"1cd782b03a0ed927139aae93ad8c9766a6af2360","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-08-02 13:37:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb351018bcf17cb8542fa4232d21f73589aa8565","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-08-02 13:43:28.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"d9664d7039843d66c62d26981d4ba77bef677272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 13:49:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2778a81a1104d89b6cd292189069190\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/54e627cf48c944b1832f85d419365a61 : SUCCESS in 3m 40s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fd4cb116975a4d608d351aa8d43925fb : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b48766bc149a67636eb996dd4dae964f8711da65","author":{"_account_id":8898,"name":"Chris Wedgwood","email":"cw@f00f.org","username":"anticw"},"date":"2022-08-09 05:18:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"04da3c689e28ba5ce5986902cdca6517963e0b22","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-11 15:53:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80f694d83e79234dafa3e833fed221306b0872fa","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-11 15:58:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2d44093219b5ab210c0660b180cde1c3d4817d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 15:58:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21c4d59230940a95c2e3a16e03d1a72404623bef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:07:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83613c0828dc4914815729cff7b73d68\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/defe24e2df8a4ffb8ded80999f847ebd : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e135426d7c297654073448ee9669b1c52ca6861c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:07:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bfcf53cf32f60aa1205ec1be33a5a3a476f2c728","revisions":{"bfcf53cf32f60aa1205ec1be33a5a3a476f2c728":{"kind":"REWORK","_number":1,"created":"2022-08-02 13:37:17.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/57/851857/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/851857/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/851857/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/851857/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/851857/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/851857/1"}}},"commit":{"parents":[{"commit":"c3e4ad50f7b9e73eb70f89341bc1876309a33087","subject":"Merge \"Temporarily turn on console logs for rocky in iweb\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c3e4ad50f7b9e73eb70f89341bc1876309a33087"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-08-02 13:30:25.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2022-08-02 13:33:32.000000000","tz":-300},"subject":"End project gating for openstack-helm-addons","message":"End project gating for openstack-helm-addons\n\nThis change adds the noop job template to the listing for\nopenstack-helm-addons as part of the effort to retire the\nrepository.\n\nChange-Id: I11fc532aab589d76525081e314fad989f244cc64\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bfcf53cf32f60aa1205ec1be33a5a3a476f2c728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bfcf53cf32f60aa1205ec1be33a5a3a476f2c728"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
