)]}'
{"id":"openstack%2Fproject-config~851956","triplet_id":"openstack%2Fproject-config~master~Icf0a2cfb1a0f5599704d6c8c9e85345d61884cd5","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-08-04 03:07:49.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"Icf0a2cfb1a0f5599704d6c8c9e85345d61884cd5","subject":"grafana: helper script updates","status":"MERGED","created":"2022-08-03 06:29:01.000000000","updated":"2022-08-04 03:26:04.000000000","submitted":"2022-08-04 03:16:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851956","meta_rev_id":"e7e424a4559517bfa4c27a52e687970c7bea7066","_number":851956,"virtual_id_number":851956,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 03:16:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"all":[{"value":1,"date":"2022-08-04 01:40:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2022-08-04 01:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-04 03:07:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-03 06:55:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 01:10:08.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-08-04 01:40:02.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"}],"messages":[{"id":"2ef44c1700f2ca0e5ec5fff0cf534b115527f39b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-03 06:29:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dea9e69d3a40e2b0a78c24ab24044651d75b74e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 06:55:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4152946952c844c8b7624fd67ab24c77\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/a540f32978334cd69a6169bc5b3bbd1c : SUCCESS in 21m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dd3a7523bdf14b3fbe704ed6d3837b8d : SUCCESS in 8m 11s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/32f46dfc1ecc4dd0bf0dffa2e2362dba : SUCCESS in 4m 09s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/f5d39882f5884e4a95f6cf56c2795a92 : SUCCESS in 13m 34s","accounts_in_message":[],"_revision_number":1},{"id":"268fb277d221e4e8fa16240899b738669c16185d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-04 00:04:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee0d138cd8d84f1d6df4be54d7c7036f8e0c44a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 00:30:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ae3934a1c3d484f85bc9f4dbc768c5a\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d796653b664641a1b258fa1258086d0a : SUCCESS in 20m 50s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/41158cefab5e449c95ea53e8063a8a08 : SUCCESS in 7m 22s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/0a16b4c05571438dae9b75bfc61c4135 : SUCCESS in 3m 51s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/7b906bba66b74099b38dd96975b4c586 : SUCCESS in 12m 00s","accounts_in_message":[],"_revision_number":2},{"id":"07e1333e1077e3e1e29f7160a86a469a8e8782fb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-04 01:10:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"274afe1b842c7a50e849875de98132c5c0f83b43","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-08-04 01:40:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5ae0c98ec83a40a471dc88977190ef8423e84428","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-04 03:07:49.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9c707aa28e4beee3a1f86dce84ce828366e7d1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:08:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5d047de9b35df87f3095f1050681c9b6a5902a70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:16:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18c00c7b901544ebaa0ae56cad2a9a1e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f6c1182f2ba9427481e054d7fec616d6 : SUCCESS in 7m 59s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/dd4fe4dc9f7146c28422e0d06593f3bc : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"bb34a958734cbaee44f15d21a4a1749088687c1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:16:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e7e424a4559517bfa4c27a52e687970c7bea7066","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:26:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6015a3b9b2c742a396a56b63b3c7210c\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/188bb5b457314648a6246dd06f8f3943 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3dbb32adb3cd497367de171a053020971f0e8788","revisions":{"e150b542b81bef32708485286af1b477e9ee01e1":{"kind":"REWORK","_number":1,"created":"2022-08-03 06:29:01.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/56/851956/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/851956/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/851956/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/851956/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/851956/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/851956/1"}}},"commit":{"parents":[{"commit":"859005d497fe14bcd7335bfd8bd95a0bdaa5e63b","subject":"grafana: add helper script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/859005d497fe14bcd7335bfd8bd95a0bdaa5e63b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-08-03 06:27:01.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-08-03 06:27:01.000000000","tz":600},"subject":"grafana: helper script updates","message":"grafana: helper script updates\n\nResponding to comments from Ibbc2b116d0c496655a7ce6bb6971e8270ac32647\n\nMake the script run in the grafana .yaml directory so you don\u0027t have\nworking-directory issues.\n\nPut the secrets in this directory and add them to .gitignore\n\nNo need for a tty on the loading call\n\nChange-Id: Icf0a2cfb1a0f5599704d6c8c9e85345d61884cd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e150b542b81bef32708485286af1b477e9ee01e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e150b542b81bef32708485286af1b477e9ee01e1"}]},"branch":"refs/heads/master"},"3dbb32adb3cd497367de171a053020971f0e8788":{"kind":"REWORK","_number":2,"created":"2022-08-04 00:04:22.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/56/851956/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/851956/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/851956/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/851956/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/851956/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/851956/2"}}},"commit":{"parents":[{"commit":"859005d497fe14bcd7335bfd8bd95a0bdaa5e63b","subject":"grafana: add helper script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/859005d497fe14bcd7335bfd8bd95a0bdaa5e63b"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-08-03 06:27:01.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-08-04 00:03:34.000000000","tz":600},"subject":"grafana: helper script updates","message":"grafana: helper script updates\n\nResponding to comments from Ibbc2b116d0c496655a7ce6bb6971e8270ac32647\n\nMake the script run in the grafana .yaml directory so you don\u0027t have\nworking-directory issues.\n\nPut the secrets in this directory and add them to .gitignore\n\nNo need for a tty on the loading call\n\nAdditionally make it executable, and have a short pause to let the\ncontainer start.\n\nChange-Id: Icf0a2cfb1a0f5599704d6c8c9e85345d61884cd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dbb32adb3cd497367de171a053020971f0e8788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dbb32adb3cd497367de171a053020971f0e8788"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
