)]}'
{"id":"openstack%2Fproject-config~851957","triplet_id":"openstack%2Fproject-config~master~I23f2b397bc7f1d8c2a959e0e90f5058cf3bf104d","project":"openstack/project-config","branch":"master","topic":"upgrade_openeuler_version","attention_set":{},"removed_from_attention_set":{"15054":{"account":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"last_update":"2022-08-04 06:50:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23f2b397bc7f1d8c2a959e0e90f5058cf3bf104d","subject":"Bump openEuler to 22.03 LTS","status":"MERGED","created":"2022-08-03 06:40:52.000000000","updated":"2022-08-04 07:34:31.000000000","submitted":"2022-08-04 06:50:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851957-upgrade_openeuler_version","meta_rev_id":"f0f4bc7716a4944a1dfd1b85c7bfff3127216222","_number":851957,"virtual_id_number":851957,"owner":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 06:50:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2022-08-03 11:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-04 06:43:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}],"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-03 06:52:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-03 08:33:06.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2022-08-03 11:39:17.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-08-04 06:43:56.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"a36208b6f33e47296e6e66c0e8efb533b737de6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2022-08-03 06:40:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fe22a8820e51ed1c9ae6cc95ef8e3e7638f4995","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 06:52:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e78d522780bd4448a78751f08b18501a\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0c25e6c1cc824b8881b4760749fae501 : SUCCESS in 6m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5177df5e490c47158882a0d546911568 : SUCCESS in 7m 46s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/8029692d2b9845d3a4c2dad7732997d5 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"bff3749d56c9992dbc6267b6dbfef53046a4c3a5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-03 11:39:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cbb79fb91331ea6713d3ec4be1e6f90c4791ee0c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-04 06:43:56.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4e5f82465e7db94e7bca507aea0cf08302545e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 06:44:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3551c8ec8007396b404effae7c419500b3f2547a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 06:50:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53876bbb739b45d799f6eccf406c653f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1c78448a3a6b4af588b0c3d87b96b8fc : SUCCESS in 6m 09s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2b58d65ae2224ebaa70c20c71df052df : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3a843fefb6922fcdeb11a655346ddab7f9858148","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 06:50:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f0f4bc7716a4944a1dfd1b85c7bfff3127216222","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 07:34:31.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38f283f463fd41528331acce5a2ef1b9\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/3369125194544ac996276a1b8fa35a7e : POST_FAILURE in 11m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30fca1376d73d430049d9ffc4b4ec69125dde20f","revisions":{"30fca1376d73d430049d9ffc4b4ec69125dde20f":{"kind":"REWORK","_number":1,"created":"2022-08-03 06:40:52.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/57/851957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/851957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/851957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/851957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/851957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/851957/1"}}},"commit":{"parents":[{"commit":"dc7b2eee15c5ec427b8cb1f106ab374bbfe09e8e","subject":"Merge \"linter: update some YAML whitespace issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dc7b2eee15c5ec427b8cb1f106ab374bbfe09e8e"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2022-08-03 06:21:12.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2022-08-03 06:40:34.000000000","tz":480},"subject":"Bump openEuler to 22.03 LTS","message":"Bump openEuler to 22.03 LTS\n\nopenEuler 20.03-LTS-SP2 was out of date in May 2022. 22.03 LTS\nis the newest LTS version. It was release in March 2022 and\nwill be maintained for 2 years. This patch upgrades the LTS\nversion. It\u0027ll be used in Devstack, Kolla-ansible and so on\nin CI jobs.\n\nChange-Id: I23f2b397bc7f1d8c2a959e0e90f5058cf3bf104d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/30fca1376d73d430049d9ffc4b4ec69125dde20f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/30fca1376d73d430049d9ffc4b4ec69125dde20f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
