)]}'
{"id":"openstack%2Fproject-config~852062","triplet_id":"openstack%2Fproject-config~master~I8a7d1577d416e4833cc5925667600f7a0540fc68","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-08-04 03:08:22.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I8a7d1577d416e4833cc5925667600f7a0540fc68","subject":"grafana: better loop for container start","status":"MERGED","created":"2022-08-04 01:17:15.000000000","updated":"2022-08-04 03:37:38.000000000","submitted":"2022-08-04 03:23:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852062","meta_rev_id":"67556ba54b75745bc929cfc4e3ba8c4e2cc85dc7","_number":852062,"virtual_id_number":852062,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 03:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2022-08-04 01:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-04 03:08:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 01:24:55.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-08-04 01:44:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"010a3add509aa54ae23554d7ac3546f27ec9256f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-04 01:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f661f07d9ac4927a92bd5778b27498992f6a172","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-04 01:24:55.000000000","message":"Patch Set 1: Code-Review+2\n\nWorst case the job runs until it hits the timeout if grafana never starts, not a big deal.","accounts_in_message":[],"_revision_number":1},{"id":"2e2786425c58ab6cb90c908d7c3f117b32b19226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 01:44:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a5b9edeeb33476381fb41b69820b30c\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b6e83a6f89c942e3ab52005f126b4210 : SUCCESS in 22m 23s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9b488957404e4bed93826a48ae042546 : SUCCESS in 10m 28s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/12504a726e754e5bafa9c8de162cb7bc : SUCCESS in 6m 30s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/8f1f854059c6401fbef1857c672afbfd : SUCCESS in 14m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f2c7ee231744dfa020a6c0b70e8405dda690f719","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-04 03:08:22.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d01e4c370e4f5cb99c949acb9251bbafc7458312","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:08:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4bfbb72c80c7a04d4026bfa8fde1f5b933867fdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:23:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c24f844fcbe4f77be1cf7b468908843\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7b37d67f85fb4a4db9f08dd04bfc2f04 : SUCCESS in 7m 58s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/4810acb028b042958d7488ca5dfdbfd2 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"44d5f9a274ab7e5e71fe7e0ebcf671e80907d151","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:23:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"67556ba54b75745bc929cfc4e3ba8c4e2cc85dc7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 03:37:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dde651eecf594f16bbd9bd3b095996cb\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/443cc038cb414801ac315de429991977 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3bebc2d0dd268a78e730776eaf24477148d3c148","revisions":{"3bebc2d0dd268a78e730776eaf24477148d3c148":{"kind":"REWORK","_number":1,"created":"2022-08-04 01:17:15.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/62/852062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/852062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/852062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/852062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/852062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/852062/1"}}},"commit":{"parents":[{"commit":"3dbb32adb3cd497367de171a053020971f0e8788","subject":"grafana: helper script updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dbb32adb3cd497367de171a053020971f0e8788"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-08-04 01:16:41.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-08-04 01:16:41.000000000","tz":600},"subject":"grafana: better loop for container start","message":"grafana: better loop for container start\n\nMake this more reliable by probing for the service\n\nChange-Id: I8a7d1577d416e4833cc5925667600f7a0540fc68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3bebc2d0dd268a78e730776eaf24477148d3c148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3bebc2d0dd268a78e730776eaf24477148d3c148"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
