)]}'
{"id":"openstack%2Fproject-config~852127","triplet_id":"openstack%2Fproject-config~master~I4ae1d93a13c47bdc5732854ab441a88e2f4e3473","project":"openstack/project-config","branch":"master","topic":"fix-nodepool","hashtags":[],"change_id":"I4ae1d93a13c47bdc5732854ab441a88e2f4e3473","subject":"Fix mismatch for openEuler image name","status":"MERGED","created":"2022-08-04 09:50:10.000000000","updated":"2022-08-04 10:52:06.000000000","submitted":"2022-08-04 10:13:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852127-fix-nodepool","meta_rev_id":"3299ff70edc12aff40581ea94062a8221031c9a8","_number":852127,"virtual_id_number":852127,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-04 10:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-08-04 09:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-08-04 09:50:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 10:00:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d37760916070729fe6c32ca8133598b9c205284","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-04 09:50:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd3062be4dfbe4d7ef76967cca6d638083e9c4a5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-04 09:50:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e979e849775e1fe49424ca459aafb8b5752f70ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:00:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a0de53a7894e15b13d4ce5ff7ea8cc\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d4646094897f4f069fc3c35dedaca3a7 : SUCCESS in 4m 20s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/20b977d3779d4ee6bddc861be708eea6 : SUCCESS in 7m 16s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/014f443dc4a44b21b932e6854ea32a00 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a67527f85fe6e237cfa5d9af48af28ddc63317c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:01:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c226745dba70347bf38238af4189fa0499fba6ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:13:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca065702b9fc432389d29819a32e58ef\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5cd0983d2bc645508d59309f0bc154f3 : SUCCESS in 11m 02s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/9328da84db674a5d9ddcdca46fe955da : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"56d713313301d23c131356a2013d9126ec1c2612","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:13:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3299ff70edc12aff40581ea94062a8221031c9a8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:52:06.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef657af7c99a4677903a013c98161514\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/73cb97a26c4b40f197161b6778757276 : POST_FAILURE in 10m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"277f30bdc6be49835ea6aba12c7703b1c5ed2412","revisions":{"277f30bdc6be49835ea6aba12c7703b1c5ed2412":{"kind":"REWORK","_number":1,"created":"2022-08-04 09:50:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/27/852127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/27/852127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/852127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/852127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/27/852127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/27/852127/1"}}},"commit":{"parents":[{"commit":"8eb686541fd880ef66037b10a10599e0779ba4f6","subject":"Merge \"Normalize projects.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8eb686541fd880ef66037b10a10599e0779ba4f6"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-08-04 09:49:33.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-08-04 09:49:33.000000000","tz":120},"subject":"Fix mismatch for openEuler image name","message":"Fix mismatch for openEuler image name\n\nOne instance was missing the -LTS suffix.\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I4ae1d93a13c47bdc5732854ab441a88e2f4e3473\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/277f30bdc6be49835ea6aba12c7703b1c5ed2412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/277f30bdc6be49835ea6aba12c7703b1c5ed2412"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
