)]}'
{"id":"openstack%2Fproject-config~852230","triplet_id":"openstack%2Fproject-config~master~I11750048841ec49c893b4c9332a6029b329b54cb","project":"openstack/project-config","branch":"master","topic":"update_openeuler_pkg_mapping","attention_set":{},"removed_from_attention_set":{"15054":{"account":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"last_update":"2022-08-05 01:45:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I11750048841ec49c893b4c9332a6029b329b54cb","subject":"Update openeEuler pkg mapping","status":"MERGED","created":"2022-08-05 01:19:02.000000000","updated":"2022-08-05 01:56:50.000000000","submitted":"2022-08-05 01:45:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852230-update_openeuler_pkg_mapping","meta_rev_id":"6ebcd96f3b2a5bf0f92ef8c465bba55138ba5c14","_number":852230,"virtual_id_number":852230,"owner":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-05 01:45:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-05 01:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-05 01:33:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-05 01:26:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-05 01:33:06.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"43c0dc0478443b11d09834ccde35a899f7743672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2022-08-05 01:19:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"999d6bbb3ed9c205d6235ef77cda344b833a64a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 01:26:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b895a81d244d4f0390dbf4a07505bfd8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/bc7728b2746b4521b6a2f589f0b40375 : SUCCESS in 3m 33s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/827514bea81c40cdaf7b4dbad2dda466 : SUCCESS in 6m 13s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/f78aefc197fa465d9de33137b8a0f275 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"48dbea58db6106d80521e1fb3b3c9b58a44cc7bb","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-05 01:33:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fd4e61bb48986b75dff888ce7b9a37a00547aac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 01:33:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"611108e979723cb6d6062b012558e7539da7d401","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 01:45:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cca031b1b374013ab86c9a0a4807cee\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d373f3296efa42e889ec038ec5154e3f : SUCCESS in 7m 44s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/cd05fcf341c548a29953f0f892b4fcc7 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c6195fe6f3e0358c04831e77d528455989ec7c33","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 01:45:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6ebcd96f3b2a5bf0f92ef8c465bba55138ba5c14","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 01:56:50.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af2286026b2849c59a25a12da440159d\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/2bd6498ce30b40e78e313c4dc9cdef53 : POST_FAILURE in 10m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e37ee42a41180b07c81307247ae40d7537fa51ed","revisions":{"e37ee42a41180b07c81307247ae40d7537fa51ed":{"kind":"REWORK","_number":1,"created":"2022-08-05 01:19:02.000000000","uploader":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"ref":"refs/changes/30/852230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/852230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/852230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/852230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/852230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/852230/1"}}},"commit":{"parents":[{"commit":"06f6f8dd59c6b04e4052f372115f150090818482","subject":"Merge \"[Grafana] Neutron dasboard change Fedora job to Centos 9 Stream\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/06f6f8dd59c6b04e4052f372115f150090818482"}]}],"author":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2022-08-05 01:16:51.000000000","tz":480},"committer":{"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","date":"2022-08-05 01:17:42.000000000","tz":480},"subject":"Update openeEuler pkg mapping","message":"Update openeEuler pkg mapping\n\nopenEuler 22.03 supports python3 only. There is no\npython-devel package.\n\nThis patch update the pkg mapping to fix the package\ninstall problem in nodepool.\n\nChange-Id: I11750048841ec49c893b4c9332a6029b329b54cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e37ee42a41180b07c81307247ae40d7537fa51ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e37ee42a41180b07c81307247ae40d7537fa51ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
