)]}'
{"id":"openstack%2Fproject-config~853863","triplet_id":"openstack%2Fproject-config~master~I4d4d70730362a8c80cbcb0b5abf57a12fad2ae9f","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-08-26 03:15:57.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-08-26 03:15:57.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-08-26 01:10:26.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I4d4d70730362a8c80cbcb0b5abf57a12fad2ae9f","subject":"Update timer settings to match docs","status":"MERGED","created":"2022-08-20 19:41:31.000000000","updated":"2022-08-26 03:15:57.000000000","submitted":"2022-08-26 03:15:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853863","meta_rev_id":"92a48112c11c9cf03075ac115a6782b650ed7973","_number":853863,"virtual_id_number":853863,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 03:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-08-26 02:46:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-26 02:50:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-26 02:50:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-20 19:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-20 20:16:40.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-08-20 22:08:23.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"43a35419225a3e304025ce60d09ddb9cc39088c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-20 19:41:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05f89e1fc56d3875f620ed3b87deeda6cdf5eb5f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-20 19:42:18.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"e6f8e6188372b2bba9ef9a7e9159db9fa7810b6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 19:49:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f520d7b5426a43799395cd10645a113b\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9e4db683d2144e8ca4a9b22716a3641c : SUCCESS in 4m 30s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/239f53564e634fd0928b3fe3df42014b : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"3f93d0b2bd171f1d828c7ac910ed4c359f01fcc5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-20 20:16:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9af46dbc7b17a6f8a0058fefc0c291a696be8596","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-20 22:08:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"342f518abbccba0ac54544a7326090c84dce6aa5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-20 22:08:31.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"61dcc25d6313fda0115a3be996460081796f9d40","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-26 01:08:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"edd6ed490c742136fc70fd4de8e25b959d3383d4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-26 01:08:59.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"1e53cd001dfde283ad465f9e2f5123b6f638a7aa","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-26 01:10:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7f98e7eb8663b953a517a0cf49d9f6bd215784af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 01:22:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/560ca72a5bdd46faa5a90778c8862087\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7044b4e243d44f3d95a8c770d6314660 : SUCCESS in 3m 27s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2482a07cd4b549f18aba4f2a99d1f137 : FAILURE in 7m 27s","accounts_in_message":[],"_revision_number":2},{"id":"7906e22b1164a048fb193e90b712a38736b97de8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-26 02:46:48.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"19e255393bf5d1a8b7af10c0d6c39edb9508faf5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-26 02:50:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"084c590fe73ff9b49be5052b8733a2d7529b13a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:00:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b133dd086b44abf81144b8ee763d78c\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1df12e04ed834dada0f0c41aa9403aa6 : SUCCESS in 6m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bb4ce25946e6404ba86dd4733a035966 : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":3},{"id":"05b58ef112e7c5ab6144ac08f9840a015395b31d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:03:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"337d81d63cb1e4f1510e85746160ca908b58bed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:15:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dad0dbb5b4854c488eaa56c549af3c57\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/640d38b3177c4ced8a2316ebd9f355ce : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":3},{"id":"92a48112c11c9cf03075ac115a6782b650ed7973","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:15:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d84a9b1ec7b625e6159366e95932f88d5e142868","revisions":{"b53c37f73b7de5ccb0fb850863a6dfa99589af62":{"kind":"REWORK","_number":1,"created":"2022-08-20 19:41:31.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/63/853863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/853863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/853863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/853863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/853863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/853863/1"}}},"commit":{"parents":[{"commit":"704503bd60c7e9f43278d3dc77277661ada63e07","subject":"Move the requirements-constraints job to the periodic-weekly pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/704503bd60c7e9f43278d3dc77277661ada63e07"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-20 19:33:51.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-20 19:33:51.000000000","tz":-300},"subject":"Update timer settings to match docs","message":"Update timer settings to match docs\n\nIn I478a64b21d9f19ece21ec5fb7a1327a106664529 (Move the\nrequirements-constraints job to the periodic-weekly pipeline) we\nswitched the job to run from the weekly pipeline.  After checking on\nSaturday at approx 17:00 UTC) the job hadn\u0027t triggered.\n\nI took this to IRC[1]:\n\n12:50 \u003ctonyb\u003e Tony Breeds:\nTotally not important but I\u0027m noticing something strange with the periodic-weekly pipeline.  Looking at: https://opendev.org/openstack/project-config/src/branch/master/zuul.d/pipelines.yaml#L214-L224 that pipeline is to run at 08:00am Saturday (UTC), but looking at https://zuul.opendev.org/t/openstack/builds?pipeline\u003dperiodic-weekly\u0026skip\u003d0 it seems to be running at 08:00am on Sunday?\n13:22 \u003cfrickler\u003e Jens Harbott\ntonyb: indeed, seems that zuul documentation is wrong. apschedule has mon\u003d0 ... sun\u003d6 https://apscheduler.readthedocs.io/en/3.x/modules/triggers/cron.html#module-apscheduler.triggers.cron\n13:23 \u003cfrickler\u003e Jens Harbott\nmost weird about it is that nobody seems to have noticed this before\n\nGiven the comment makes sense lets update the code to match.\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-infra/%23openstack-infra.2022-08-20.log.html#t2022-08-20T17:50:48\n\nChange-Id: I4d4d70730362a8c80cbcb0b5abf57a12fad2ae9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b53c37f73b7de5ccb0fb850863a6dfa99589af62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b53c37f73b7de5ccb0fb850863a6dfa99589af62"}]},"branch":"refs/heads/master"},"a9e2cb1b6fc415a23af47a638739b6c86d14449b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-26 01:08:40.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/63/853863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/853863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/853863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/853863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/853863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/853863/2"}}},"commit":{"parents":[{"commit":"d67d326bef6ab90af3302fbe314e176c2623ddc6","subject":"Merge \"Formally retire puppet-tuskar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d67d326bef6ab90af3302fbe314e176c2623ddc6"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-20 19:33:51.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-26 01:08:30.000000000","tz":-420},"subject":"Update timer settings to match docs","message":"Update timer settings to match docs\n\nIn I478a64b21d9f19ece21ec5fb7a1327a106664529 (Move the\nrequirements-constraints job to the periodic-weekly pipeline) we\nswitched the job to run from the weekly pipeline.  After checking on\nSaturday at approx 17:00 UTC) the job hadn\u0027t triggered.\n\nI took this to IRC[1]:\n\n12:50 \u003ctonyb\u003e Tony Breeds:\nTotally not important but I\u0027m noticing something strange with the periodic-weekly pipeline.  Looking at: https://opendev.org/openstack/project-config/src/branch/master/zuul.d/pipelines.yaml#L214-L224 that pipeline is to run at 08:00am Saturday (UTC), but looking at https://zuul.opendev.org/t/openstack/builds?pipeline\u003dperiodic-weekly\u0026skip\u003d0 it seems to be running at 08:00am on Sunday?\n13:22 \u003cfrickler\u003e Jens Harbott\ntonyb: indeed, seems that zuul documentation is wrong. apschedule has mon\u003d0 ... sun\u003d6 https://apscheduler.readthedocs.io/en/3.x/modules/triggers/cron.html#module-apscheduler.triggers.cron\n13:23 \u003cfrickler\u003e Jens Harbott\nmost weird about it is that nobody seems to have noticed this before\n\nGiven the comment makes sense lets update the code to match.\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-infra/%23openstack-infra.2022-08-20.log.html#t2022-08-20T17:50:48\n\nChange-Id: I4d4d70730362a8c80cbcb0b5abf57a12fad2ae9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a9e2cb1b6fc415a23af47a638739b6c86d14449b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a9e2cb1b6fc415a23af47a638739b6c86d14449b"}]},"branch":"refs/heads/master"},"d84a9b1ec7b625e6159366e95932f88d5e142868":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-26 02:46:48.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/63/853863/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/853863/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/853863/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/853863/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/853863/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/853863/3"}}},"commit":{"parents":[{"commit":"46838fbda8faf1e203a929b057305d9571db0afe","subject":"Exclude ansible-lint 6.5.1 also.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/46838fbda8faf1e203a929b057305d9571db0afe"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-20 19:33:51.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-26 02:46:39.000000000","tz":-420},"subject":"Update timer settings to match docs","message":"Update timer settings to match docs\n\nIn I478a64b21d9f19ece21ec5fb7a1327a106664529 (Move the\nrequirements-constraints job to the periodic-weekly pipeline) we\nswitched the job to run from the weekly pipeline.  After checking on\nSaturday at approx 17:00 UTC) the job hadn\u0027t triggered.\n\nI took this to IRC[1]:\n\n12:50 \u003ctonyb\u003e Tony Breeds:\nTotally not important but I\u0027m noticing something strange with the periodic-weekly pipeline.  Looking at: https://opendev.org/openstack/project-config/src/branch/master/zuul.d/pipelines.yaml#L214-L224 that pipeline is to run at 08:00am Saturday (UTC), but looking at https://zuul.opendev.org/t/openstack/builds?pipeline\u003dperiodic-weekly\u0026skip\u003d0 it seems to be running at 08:00am on Sunday?\n13:22 \u003cfrickler\u003e Jens Harbott\ntonyb: indeed, seems that zuul documentation is wrong. apschedule has mon\u003d0 ... sun\u003d6 https://apscheduler.readthedocs.io/en/3.x/modules/triggers/cron.html#module-apscheduler.triggers.cron\n13:23 \u003cfrickler\u003e Jens Harbott\nmost weird about it is that nobody seems to have noticed this before\n\nGiven the comment makes sense lets update the code to match.\n\n[1] https://meetings.opendev.org/irclogs/%23openstack-infra/%23openstack-infra.2022-08-20.log.html#t2022-08-20T17:50:48\n\nChange-Id: I4d4d70730362a8c80cbcb0b5abf57a12fad2ae9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d84a9b1ec7b625e6159366e95932f88d5e142868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d84a9b1ec7b625e6159366e95932f88d5e142868"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
