)]}'
{"id":"openstack%2Fproject-config~853893","triplet_id":"openstack%2Fproject-config~master~I64fac34945ea5c6ec91ddd442281fcaba2c53271","project":"openstack/project-config","branch":"master","topic":"fedora-36","hashtags":[],"change_id":"I64fac34945ea5c6ec91ddd442281fcaba2c53271","subject":"nodepool: Add Fedora 36","status":"MERGED","created":"2022-08-22 01:26:36.000000000","updated":"2022-08-22 03:09:01.000000000","submitted":"2022-08-22 02:02:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853893-fedora-36","meta_rev_id":"adc10ca12e046d6b8ba5ff015471d0029f1dd91e","_number":853893,"virtual_id_number":853893,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 02:02:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-22 01:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-22 01:52:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-22 01:37:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dddf81472c2dd955991143e4ac4be414bf19a9b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-22 01:26:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a13691ee0a1ad0155111ea632b18fb94a30c6403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 01:37:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b98804539f1e4af29a1adf888dc859d8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/567b7b85b8d844c1a3c560cbd38b795e : SUCCESS in 3m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c4ef5869a326484eafbc86f6b666c8ab : SUCCESS in 6m 50s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/09934a7e5f4c4f2caeb0e1a3f8700f1a : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"25842c7458468a68c771e3921e8a1a08fa974adb","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-08-22 01:52:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c493069a8865d6dfa38795a5f70853960dbb86a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 01:53:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"49b9e7388911c220245eb3c0d5f5d2ded7ac210f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 02:02:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbb6fb13677244f6b9a14b3fbddc4a44\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5662ab771afb4a889b749110b5e5e282 : SUCCESS in 7m 10s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/9245b5f1550a411b815462409cd0b959 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"afe92e30fa91aa49dd24b2f3fb248a3949163b7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 02:02:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"adc10ca12e046d6b8ba5ff015471d0029f1dd91e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 03:09:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a05d6bc5d5014b488f4bd7a8f733a4c3\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/5e549fc175b44d2897aafa30759d236d : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36b9c302e5f0e5e8305ce61d455e914788a76c29","revisions":{"36b9c302e5f0e5e8305ce61d455e914788a76c29":{"kind":"REWORK","_number":1,"created":"2022-08-22 01:26:36.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/93/853893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/853893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/853893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/853893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/853893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/853893/1"}}},"commit":{"parents":[{"commit":"704503bd60c7e9f43278d3dc77277661ada63e07","subject":"Move the requirements-constraints job to the periodic-weekly pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/704503bd60c7e9f43278d3dc77277661ada63e07"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-08-22 01:24:39.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-08-22 01:25:09.000000000","tz":600},"subject":"nodepool: Add Fedora 36","message":"nodepool: Add Fedora 36\n\nAdd Fedora 36 builds\n\nChange-Id: I64fac34945ea5c6ec91ddd442281fcaba2c53271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36b9c302e5f0e5e8305ce61d455e914788a76c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36b9c302e5f0e5e8305ce61d455e914788a76c29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
