)]}'
{"id":"openstack%2Fproject-config~854431","triplet_id":"openstack%2Fproject-config~master~I5c8c8350173d0aee0683c7fe40fafce7b5602a51","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2022-09-01 08:08:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c8c8350173d0aee0683c7fe40fafce7b5602a51","subject":"Add nested-virt-rockylinux-9","status":"MERGED","created":"2022-08-24 15:47:13.000000000","updated":"2022-09-01 08:38:36.000000000","submitted":"2022-09-01 08:08:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854431","meta_rev_id":"37d31c76832ddc5f80c56f8476df2999e0d41368","_number":854431,"virtual_id_number":854431,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-01 08:08:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-09-01 08:01:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-24 15:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-09-01 08:01:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-24 15:56:44.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-08-24 15:59:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-01 08:01:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ae5e52cb5fff20381fed1b6121145d0d8e68e2da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-24 15:47:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5c03a799f916175529fbdb54f9728fba8a6f38f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-24 15:56:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2fcd277945fbc3675791f5e30fcde96a1be16b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 15:59:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/721abbabcbfe48c2acbea25dc5a5b37b\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e1161e8b34694932bdb7c2e2f6fad13f : SUCCESS in 7m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/008e0fac68224d6783d378336ebd13f9 : FAILURE in 9m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/d88c0a24bf6a410987f07f23f547c948 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"667ad0309a640306d588f70830fef08f89a58a80","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-24 16:13:50.000000000","message":"Patch Set 1:\n\nThe linters failure is unrelated, looks like new ansiblelint has decided to grace us with a visit. Should be fixed shortly.","accounts_in_message":[],"_revision_number":1},{"id":"25a72b270bffad5637e047b47aad1d192d0d25ae","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-24 18:51:45.000000000","message":"Patch Set 1:\n\nrecheck now that we\u0027re skipping ansible-lint 6.5.0","accounts_in_message":[],"_revision_number":1},{"id":"2325c76ccc036b730798f5c05937b6f4e6e8df94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-24 18:59:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1c2382772714bcda5c517cd065c9b06\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/3a3fe3ca2cd54e52a97e0249ae92c1bd : SUCCESS in 3m 35s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e1cee125408840f2b98325229f95d0a3 : SUCCESS in 6m 18s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/1d153c547c6f4ff8881855d5dd612708 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"cf7f38e1ecd26a4077d5db722ce3f28c857698c6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-01 08:01:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51cd63b90e2f5749e64103fc011c42df87964d9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:01:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29993fd6eb911fddfa710fb9e7e232b1bcc6dc51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:08:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/202eeacdce5b4e48b964db0d9534ee60\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/43f8dbd7dfd94bbea3c97876afc54d70 : SUCCESS in 5m 49s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/f07c4ffb27744c3b88d565a8bc2316a4 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"52ec4d1f203bf425a02fb51710ab3bbe373fb2a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:08:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"37d31c76832ddc5f80c56f8476df2999e0d41368","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 08:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d132017388f84781b50ad8299b466b24\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/efb5a5ccc8c74edc8b4efc47563065a5 : SUCCESS in 11m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"13552d20e5b9c6e184fe8922f66cc07bb61eb3ed","revisions":{"13552d20e5b9c6e184fe8922f66cc07bb61eb3ed":{"kind":"REWORK","_number":1,"created":"2022-08-24 15:47:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/31/854431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/854431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/854431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/854431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/854431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/854431/1"}}},"commit":{"parents":[{"commit":"36b9c302e5f0e5e8305ce61d455e914788a76c29","subject":"nodepool: Add Fedora 36","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36b9c302e5f0e5e8305ce61d455e914788a76c29"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-24 15:46:50.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2022-08-24 15:47:11.000000000","tz":120},"subject":"Add nested-virt-rockylinux-9","message":"Add nested-virt-rockylinux-9\n\nChange-Id: I5c8c8350173d0aee0683c7fe40fafce7b5602a51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/13552d20e5b9c6e184fe8922f66cc07bb61eb3ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/13552d20e5b9c6e184fe8922f66cc07bb61eb3ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
