)]}'
{"id":"openstack%2Fproject-config~854701","triplet_id":"openstack%2Fproject-config~master~I641a1246f870869153df782381fd9dafa180da26","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-08-26 03:15:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I641a1246f870869153df782381fd9dafa180da26","subject":"Exclude ansible-lint 6.5.1 also.","status":"MERGED","created":"2022-08-26 02:46:48.000000000","updated":"2022-08-26 03:18:39.000000000","submitted":"2022-08-26 03:15:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854701","meta_rev_id":"95e1c476709174bfd2c3ca7c73d7c8b1bcacf416","_number":854701,"virtual_id_number":854701,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 03:15:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-26 02:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-26 02:47:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 02:47:24.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-08-26 03:02:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1383b6b3a000b85582de2d62f6639ac08b05df4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-26 02:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48fa83d1c9c5b5464c0feb17c251e1cf5de2cb83","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-08-26 02:47:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee899b0b7fa735e07a9cf90a3f621b55b3aa5536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:02:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/710044bd1b514b7690fc3db7a055b68f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8303d699f2134472a38b08eb79351d18 : SUCCESS in 7m 54s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/b12645b1b4474f8ab90de544739ae6af : SUCCESS in 7m 23s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ac10a9449355444987cbf991d65c40f4 : SUCCESS in 10m 37s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/c0a04f8103fe437f8e35ec175e9c1688 : SUCCESS in 6m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b3b563c891214abcb48a8f7630bfa5c6 : SUCCESS in 6m 37s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/976cba2267c846489d26a0befc7b1370 : SUCCESS in 4m 48s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c96a5303a88f4acf83f0951db98f25d7 : SUCCESS in 7m 01s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/c34bdfbe98f34c7da89d64782988061b : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"ce7d4e1679b8e989d5608a8330fea8b22224621b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:02:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfeddf0f89d825e731b44b6aa70e1f4accc26883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:15:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fa9d9eb699e43e4b2fa8ad8e0b4f1e7\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/f034113d06514b0bb8fcdf460c3e9ed7 : SUCCESS in 3m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9b280f9bf6654a8bbca2d29546504c8f : SUCCESS in 10m 59s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/f60ce5cb95514e7caa0371854c24076b : SUCCESS in 3m 44s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/02a12badefae4a7d8bf1bdeb288fca34 : SUCCESS in 4m 13s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/a82eb1e42d1149698d8036403aac82b8 : SUCCESS in 3m 55s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/26b0a7c141d04633ac408ebd534c1fcb : SUCCESS in 6m 27s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/16f38120a12e44aeb01dd7f4bba47a61 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"db47250b553140cdddef085e2084ad6ded5f43c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:15:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"95e1c476709174bfd2c3ca7c73d7c8b1bcacf416","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 03:18:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb8037a13bb34cb7b471fbfe416c8ae9\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/d5d9a39b93ae41e3b753a704e4825b70 : SUCCESS in 55s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/66d6373e13804df2afcf321923477007 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46838fbda8faf1e203a929b057305d9571db0afe","revisions":{"46838fbda8faf1e203a929b057305d9571db0afe":{"kind":"REWORK","_number":1,"created":"2022-08-26 02:46:48.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/01/854701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/854701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/854701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/854701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/854701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/854701/1"}}},"commit":{"parents":[{"commit":"d67d326bef6ab90af3302fbe314e176c2623ddc6","subject":"Merge \"Formally retire puppet-tuskar\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d67d326bef6ab90af3302fbe314e176c2623ddc6"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-26 02:44:26.000000000","tz":-420},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-26 02:46:39.000000000","tz":-420},"subject":"Exclude ansible-lint 6.5.1 also.","message":"Exclude ansible-lint 6.5.1 also.\n\nansible-lint introduced a regression as described in\n\"Don\u0027t use ansible-lint 6.5.0\" I1770e034772b55b993d276d580c5a75d059bd822\nthe fix is still in draft and 6.5.1 has been released so lets skip that\ntoo\n\nChange-Id: I641a1246f870869153df782381fd9dafa180da26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/46838fbda8faf1e203a929b057305d9571db0afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/46838fbda8faf1e203a929b057305d9571db0afe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
