)]}'
{"id":"openstack%2Fproject-config~858523","triplet_id":"openstack%2Fproject-config~master~Ie157ab7f667dbdc313b7fdbfc1df71874e8bd0fc","project":"openstack/project-config","branch":"master","topic":"rockylinux-openeuler","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-09-20 14:20:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie157ab7f667dbdc313b7fdbfc1df71874e8bd0fc","subject":"Add missing rockylinux and openeuler entries","status":"MERGED","created":"2022-09-20 12:35:22.000000000","updated":"2022-09-20 14:54:39.000000000","submitted":"2022-09-20 14:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858523-rockylinux-openeuler","meta_rev_id":"aa4e0a8df4ae492ec69f0c4fd0b29926c61c710e","_number":858523,"virtual_id_number":858523,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-20 14:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-09-20 14:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2022-09-20 12:52:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2022-09-20 14:08:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-20 12:45:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-20 12:52:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-09-20 14:08:59.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"dc03d3a1a2e730d37ebcd4b3ff5da0475bcebfa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-09-20 12:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c33fbc261d3d6f7c0eecbc0369f2baae6cb9f18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 12:45:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38e53776513348c5b50e4a82d05eeed8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0352b8a2e5cc4c73a7daf166445c3dfe : SUCCESS in 3m 35s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/932c87895b24476ebb9e478f905caf81 : SUCCESS in 7m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/aedea9b424f14a62940e562222b114ea : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"bf9e1749758bfcbaeeb7a7e79ac4b8547fa80911","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-20 12:52:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a78ef1d3029f8b815b3d98ab5d7a087bed76cbaf","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-09-20 14:08:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e06c51fd7a9ed9477932c4336e83f86a28ae982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 14:09:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f16ba19420a48e983d9d034983bfb8db52df297","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 14:20:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/710651a0771c4c449fb621337eb737a2\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/542b9ee3fb464c04b0bf8a6c0a860000 : SUCCESS in 6m 14s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/7047197e6aa542fda73233176f993fbf : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ca4bd30a269e8e292ee02df56ce311aebe24b71e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 14:20:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aa4e0a8df4ae492ec69f0c4fd0b29926c61c710e","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 14:54:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1abc566fa9048fea85d44d556211a4e\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/95f34f53c939449bb99c04bdc639dc41 : SUCCESS in 11m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcd02797eeb57ff981edaf77872033eee3c19d7f","revisions":{"fcd02797eeb57ff981edaf77872033eee3c19d7f":{"kind":"REWORK","_number":1,"created":"2022-09-20 12:35:22.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/23/858523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/858523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/858523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/858523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/858523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/858523/1"}}},"commit":{"parents":[{"commit":"b597a94b2e8886087a21d0d3fe7dd86f5323f06c","subject":"Add missing fedora-36 label to nl01","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b597a94b2e8886087a21d0d3fe7dd86f5323f06c"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-09-20 12:30:21.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-09-20 12:35:16.000000000","tz":0},"subject":"Add missing rockylinux and openeuler entries","message":"Add missing rockylinux and openeuler entries\n\nOur various nodepool launcher configs have gotten out of sync with\nrecent additions. Try to match them up better by adding rockylinux\nand openeuler to places where they were missing. In particular, this\naddresses an inconsistency on nl04 which caused the service to crash\nat last restart due to rockylinux missing from the main diskimages\nlist.\n\nChange-Id: Ie157ab7f667dbdc313b7fdbfc1df71874e8bd0fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcd02797eeb57ff981edaf77872033eee3c19d7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcd02797eeb57ff981edaf77872033eee3c19d7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
