)]}'
{"id":"openstack%2Fproject-config~859977","triplet_id":"openstack%2Fproject-config~master~Ib633a621fbf004f2a34b8826014e3625c093c21f","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2022-11-23 09:07:38.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2022-11-23 09:07:38.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-10-24 16:02:10.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ib633a621fbf004f2a34b8826014e3625c093c21f","subject":"Add post-review pipeline","status":"MERGED","created":"2022-09-30 11:42:57.000000000","updated":"2022-11-23 09:07:38.000000000","submitted":"2022-11-23 09:07:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"859977","meta_rev_id":"9e45aebe5b01e9c5c61dd1d1dacd94ee75edd2da","_number":859977,"virtual_id_number":859977,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-23 09:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-10-24 17:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2022-11-23 08:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2022-11-23 08:56:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-30 11:53:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-30 12:21:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2022-10-10 07:35:58.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"CC"},{"updated":"2022-10-24 15:16:44.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2022-10-24 17:03:40.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-11-23 08:51:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"5db471adec93c167dd2dc564a0c36a00cb7bda5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-09-30 11:42:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b2a18a2b5b02d17285e3ca3ba4a05637100b3b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 11:53:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f5ccb5e15e4ddca9ed8fea7d0fbc16\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/528702840a3c45abac259b3cb7c094a1 : SUCCESS in 6m 39s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d58c7a5889084c3cb74609c69022f903 : SUCCESS in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0527c06e35c74665b7468777c18a65b89e1145cb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-09-30 12:21:20.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"eb9d06d89c41d60f8fdd73c80df8cf63e1d4161a","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-09-30 12:35:39.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d2ef4030aa08170d3147da0dcae3f358d03479c4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-09-30 12:57:05.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2271e38e765c9e887a36855292b0f3f512dbac4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-09-30 13:07:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b981a7fb83c4bab7bb3b8361df982697374b374","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-09-30 13:08:53.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4cb1763a7496110d7913d8d37acd5577523cd5b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-30 13:33:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bda4c3e31c154fa389b5bd23109f78e8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1683d9ab8112401d91f3f675c596ad45 : SUCCESS in 4m 00s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f71cbff25183421c852f59222c4557ea : SUCCESS in 8m 22s","accounts_in_message":[],"_revision_number":2},{"id":"f850fce594b5380a183afae5140295a41630132d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-10-10 07:35:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63bbe9a3619ccc2eb00e6b42426d7e8a47881af4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-10 07:44:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"88e941b2ad260925be6ec42fa93885cd1880e9ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-10 07:56:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d804a7e90ad84e0a80027fb20d459998\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d2299c9f100d48eca7d2812629542af9 : SUCCESS in 2m 55s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/714b612a191946caaf880ce025aad2fc : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":3},{"id":"17515dc88c180afc206907ecccf8800252c31b73","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-10-24 15:16:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6aeaf9b4a50a2450785d011dc2683d57b160e894","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-24 16:02:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d6d26efb2b9544f44a82e9e9e65d943c9c645d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-24 16:53:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"39a53b7e804a804ead311bde5c023b893cad37d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 16:54:02.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/project-config for 859977,4","accounts_in_message":[],"_revision_number":4},{"id":"fa4774c33a2af865801cabc0279eaddf984782b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2022-10-24 17:02:21.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"fea5c6294a592d3e7da7cc705dc9b72fbd455f04","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-10-24 17:03:40.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0d2207430d6845f3fcaa76fcfc57185f3686053a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 17:15:39.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ce2eae6c5ee44ac91b80d9a9ad7ee15\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/28571d47df2e4ab89ca3ac31646da2f2 : SUCCESS in 2m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/be72cf846d4f482cb48fe767cb7cda0b : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":5},{"id":"02228d1066c9312d82b0bbcdcf75d0af16c5c9ef","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-23 08:51:15.000000000","message":"Patch Set 5: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9cdad6a671a19a5dfba85587580d13bc4c4acaf8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-23 08:56:19.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"0c035693be02a337e50f5b72cb5d5c58a0adc0ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 08:56:48.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"38db6dd28681aaf3dc6eb235326d291609906241","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 09:07:38.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f03046f7dcf145419a7a909d3284a977\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ca0911f3f7c74a108e05434b6a3e0f7e : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":5},{"id":"9e45aebe5b01e9c5c61dd1d1dacd94ee75edd2da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 09:07:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"51e501abc49865589ea1d055ae265b5047beea17","revisions":{"03a365cc373e330171d1d0092ea43023350ac7d3":{"kind":"REWORK","_number":1,"created":"2022-09-30 11:42:57.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/77/859977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/859977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/859977/1"}}},"commit":{"parents":[{"commit":"68fa766c5ed058030104eb9211c48ae06804b610","subject":"Add Post-Check flag to OpenStackSDK project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/68fa766c5ed058030104eb9211c48ae06804b610"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-09-30 11:34:59.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-09-30 11:42:31.000000000","tz":120},"subject":"Add check-post pipeline","message":"Add check-post pipeline\n\nIn order to start using jobs with secrets before gating it required to\nhave a special post review pipeline. This pipeline can be triggered by\nsetting Check-Post gerrit flag.\n\nTo address potential situation where check-post jobs pass, but check\njobs fails it only sets Verified to -1 once jobs fail.\n\nFor the proper setup jobs of the check-post pipeline should also appear\nin gate pipeline.\n\nChange-Id: Ib633a621fbf004f2a34b8826014e3625c093c21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03a365cc373e330171d1d0092ea43023350ac7d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03a365cc373e330171d1d0092ea43023350ac7d3"}]},"branch":"refs/heads/master"},"accc9d832c28c635c3ae9ae68ec828fd4e5e0a3b":{"kind":"REWORK","_number":2,"created":"2022-09-30 13:07:56.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/77/859977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/859977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/859977/2"}}},"commit":{"parents":[{"commit":"d6badfd4c54910b0e94fe2f769e0964ff0692512","subject":"Add Allow-Post-Review flag to OpenStackSDK project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6badfd4c54910b0e94fe2f769e0964ff0692512"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-09-30 11:34:59.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-09-30 13:06:57.000000000","tz":120},"subject":"Add post-review pipeline","message":"Add post-review pipeline\n\nIn order to start using jobs with secrets before gating it required to\nhave a special post review pipeline. This pipeline can be triggered by\nsetting Allow-Post-Review gerrit flag.\n\nTo address potential races for Verified flag this pipeline does not set\nit.\n\nFor the proper setup jobs of the post-review pipeline should also appear\nin gate pipeline.\n\nChange-Id: Ib633a621fbf004f2a34b8826014e3625c093c21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/accc9d832c28c635c3ae9ae68ec828fd4e5e0a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/accc9d832c28c635c3ae9ae68ec828fd4e5e0a3b"}]},"branch":"refs/heads/master"},"57df82c76b7f50a6a520896135830ad52d2ab1d8":{"kind":"REWORK","_number":3,"created":"2022-10-10 07:44:47.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/77/859977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/859977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/859977/3"}}},"commit":{"parents":[{"commit":"d6badfd4c54910b0e94fe2f769e0964ff0692512","subject":"Add Allow-Post-Review flag to OpenStackSDK project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6badfd4c54910b0e94fe2f769e0964ff0692512"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-09-30 11:34:59.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-10 07:44:12.000000000","tz":120},"subject":"Add post-review pipeline","message":"Add post-review pipeline\n\nIn order to start using jobs with secrets before gating it required to\nhave a special post review pipeline. This pipeline can be triggered by\nsetting Allow-Post-Review gerrit flag.\n\nTo address potential races for Verified flag this pipeline does not set\nit.\n\nFor the proper setup jobs of the post-review pipeline should also appear\nin gate pipeline.\n\nChange-Id: Ib633a621fbf004f2a34b8826014e3625c093c21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57df82c76b7f50a6a520896135830ad52d2ab1d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/57df82c76b7f50a6a520896135830ad52d2ab1d8"}]},"branch":"refs/heads/master"},"c4aadaa1aa4e06d942f58cec9c6a6bf5bc62509f":{"kind":"REWORK","_number":4,"created":"2022-10-24 16:53:38.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/77/859977/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/859977/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/859977/4"}}},"commit":{"parents":[{"commit":"d6badfd4c54910b0e94fe2f769e0964ff0692512","subject":"Add Allow-Post-Review flag to OpenStackSDK project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d6badfd4c54910b0e94fe2f769e0964ff0692512"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-09-30 11:34:59.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-24 16:53:13.000000000","tz":120},"subject":"Add post-review pipeline","message":"Add post-review pipeline\n\nIn order to start using jobs with secrets before gating it required to\nhave a special post review pipeline. This pipeline can be triggered by\nsetting Allow-Post-Review gerrit flag.\n\nTo address potential races for Verified flag this pipeline does not set\nit.\n\nFor the proper setup jobs of the post-review pipeline should also appear\nin gate pipeline.\n\nChange-Id: Ib633a621fbf004f2a34b8826014e3625c093c21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c4aadaa1aa4e06d942f58cec9c6a6bf5bc62509f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c4aadaa1aa4e06d942f58cec9c6a6bf5bc62509f"}]},"branch":"refs/heads/master"},"51e501abc49865589ea1d055ae265b5047beea17":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-10-24 17:02:21.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/77/859977/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/859977/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/859977/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/859977/5"}}},"commit":{"parents":[{"commit":"ef92cc76956f7fdcfc14a41f6bf6cf93a7cbb7e6","subject":"Add Allow-Post-Review flag to OpenStackSDK project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ef92cc76956f7fdcfc14a41f6bf6cf93a7cbb7e6"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-09-30 11:34:59.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2022-10-24 17:02:21.000000000","tz":0},"subject":"Add post-review pipeline","message":"Add post-review pipeline\n\nIn order to start using jobs with secrets before gating it required to\nhave a special post review pipeline. This pipeline can be triggered by\nsetting Allow-Post-Review gerrit flag.\n\nTo address potential races for Verified flag this pipeline does not set\nit.\n\nFor the proper setup jobs of the post-review pipeline should also appear\nin gate pipeline.\n\nChange-Id: Ib633a621fbf004f2a34b8826014e3625c093c21f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/51e501abc49865589ea1d055ae265b5047beea17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/51e501abc49865589ea1d055ae265b5047beea17"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[]}
