)]}'
{"id":"openstack%2Fproject-config~860094","triplet_id":"openstack%2Fproject-config~master~Ie0cfc69eccc4238b8b20250ec2cda0438ea076a4","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2022-10-30 18:09:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0cfc69eccc4238b8b20250ec2cda0438ea076a4","subject":"Add rockylinux 9 to OSA grafana","status":"MERGED","created":"2022-10-03 14:19:08.000000000","updated":"2022-10-30 18:37:36.000000000","submitted":"2022-10-30 18:09:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860094","meta_rev_id":"5d1100e57673c50c0c8dda01db5ea76e78020146","_number":860094,"virtual_id_number":860094,"owner":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-30 18:09:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2022-10-30 09:24:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-03 16:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2022-10-30 09:24:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-03 14:20:21.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"},{"updated":"2022-10-03 14:20:21.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-10-03 14:49:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-03 16:35:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-10-30 09:24:33.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"ce36fbc3981b21f759497a75c895a89f95e16fc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-10-03 14:19:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7c03e3605dd615a41c50f371c17c2afedbae02e","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2022-10-03 14:20:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b05ab142df7067c466df5b33224c4d2901515b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-03 14:49:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85d8e83ba8c04994a2ed0a55df84b1d6\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d3773e5c97444c70979535382daf1f1d : SUCCESS in 24m 31s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0905c49446b646329b6dad1225e304b6 : SUCCESS in 7m 14s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/249484acac554f2bbe39c9eee181b9a1 : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":1},{"id":"48a7248b57080c44a67d3880bddd35070556150f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-03 16:35:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bd2da1b34b84c148dcce173ac9e8b1e526f48cf3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-10-30 09:24:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f139c82f378ee5341dfb72af49b98e8ca8dcfef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 09:24:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2112fab33ec9fdeec6af2fcdea695de093de22e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 09:48:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/452391c936434621a433d7c04926572b\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/aa6d2f6c64e644f483b79a771df25d74 : SUCCESS in 19m 30s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d8e1390fde9948759e0f8c155cf429f3 : RETRY_LIMIT in 2m 13s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/7affeaca55c8418cb50565563b1dda7d : SUCCESS in 11m 48s","accounts_in_message":[],"_revision_number":1},{"id":"5bfe4cbe23015b83fa982432f2741237594a8528","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-30 12:10:08.000000000","message":"Patch Set 1:\n\nIdb717f2e45a09c61523ca96382eb3a275700074d should solve the job failures.","accounts_in_message":[],"_revision_number":1},{"id":"37dec6252d3571fff309e1f27b5d82c37394e658","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-30 17:37:07.000000000","message":"Patch Set 1:\n\nrecheck now that jobs should be passing again","accounts_in_message":[],"_revision_number":1},{"id":"078fd9f7f5a728ced21e5f7d3b355a925a8e45f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 17:54:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48f809f7ea124ae5a9902e4d166cb3ab\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2064f436bef041aca7398c4fcf4e2ec6 : SUCCESS in 16m 26s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/20b8fa42bb42435f9afb3e6425fa50f3 : SUCCESS in 4m 36s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/9359aeb4debf46b8b14e7db3ba3361ba : SUCCESS in 12m 33s","accounts_in_message":[],"_revision_number":1},{"id":"1e3c5a16508c32328f5f055ca6df8beda386cf8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 17:54:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f36041de88a0692c8ab3623493a06bf06c82eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 18:09:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3ff245ebe00460c8b2c34e26359dd6b\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/747d50b98d2d4f899cf5b889d4a3c107 : SUCCESS in 13m 53s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d4e0c21ce96a4233a004f31ff082fcbc : SUCCESS in 4m 31s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/ac0598c5f85940e6b7e33e42de2fa2ec : SUCCESS in 9m 57s","accounts_in_message":[],"_revision_number":1},{"id":"bd3a443ec6f57f6d8bcc5afe1b58943452471ee5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 18:09:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5d1100e57673c50c0c8dda01db5ea76e78020146","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 18:37:36.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd3d841f83d0413098b66262ae6606a6\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/d1f67b42388c41a0abcbae702767be6a : RETRY_LIMIT in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d39b89c486b5f50b43ab730836a62d11ae1459f9","revisions":{"d39b89c486b5f50b43ab730836a62d11ae1459f9":{"kind":"REWORK","_number":1,"created":"2022-10-03 14:19:08.000000000","uploader":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"ref":"refs/changes/94/860094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/860094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/860094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/860094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/860094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/860094/1"}}},"commit":{"parents":[{"commit":"41686c0bee0dd787c785b8c43c3a07cebb97d991","subject":"Merge \"charms: add new ceph subteam\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41686c0bee0dd787c785b8c43c3a07cebb97d991"}]}],"author":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-10-03 14:19:03.000000000","tz":-240},"committer":{"name":"Neil Hanlon","email":"neil@shrug.pw","date":"2022-10-03 14:19:03.000000000","tz":-240},"subject":"Add rockylinux 9 to OSA grafana","message":"Add rockylinux 9 to OSA grafana\n\nChange-Id: Ie0cfc69eccc4238b8b20250ec2cda0438ea076a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d39b89c486b5f50b43ab730836a62d11ae1459f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d39b89c486b5f50b43ab730836a62d11ae1459f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[]}
