)]}'
{"id":"openstack%2Fproject-config~860474","triplet_id":"openstack%2Fproject-config~master~I61493e74d9955a444326d72e085f511a838f88dd","project":"openstack/project-config","branch":"master","topic":"ansible-6","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-10-06 16:21:11.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}},"hashtags":[],"change_id":"I61493e74d9955a444326d72e085f511a838f88dd","subject":"Update OpenDev tenants\u0027 default ansible to 6","status":"MERGED","created":"2022-10-05 15:12:52.000000000","updated":"2022-10-06 16:58:21.000000000","submitted":"2022-10-06 16:31:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860474-ansible-6","meta_rev_id":"16f33a5f3f7f584b602e2fff4bd8891a6abb900f","_number":860474,"virtual_id_number":860474,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-06 16:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2022-10-05 16:20:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-05 15:59:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2022-10-06 16:21:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-05 15:26:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-05 15:59:06.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-10-05 16:20:16.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"}],"messages":[{"id":"c038ba91d7c70d32546366b2395aaad1f6f056a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-10-05 15:12:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"602c528fda90c25443a375f9a4229f38e0f95473","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-05 15:26:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b32ffed931a43258601f86c8cdf4400\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/349a4b20ac4945b4b563408810d62b7f : SUCCESS in 4m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d8e03ae77b314851bba0cef036b19cda : SUCCESS in 7m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/32262d0a493b404587d9cdda848eb44b : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8dc16d06ffcc6b7ad5f4ad54a3872c0c694f4f28","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-05 15:59:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"19057a09dfc32567ceef3ef21d89fd6a69a42604","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2022-10-05 16:20:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cc713506f8f47b9d271ecdbfba3646be187b92a9","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-10-06 16:21:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"39b057d19826c35fe7183a680b1e3838b2495103","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:21:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"918a0dd7b67ac60d55b7243cce8e90163bd8478c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:31:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb2874d8a863458193bbd852e865bbc9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7ea7453f755746069a63a6cf42ace01d : SUCCESS in 6m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b5a5347f5b2e4cdca94a15031b02c3b1 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b876f8b31a2e94c79c09d300508275a722d8e192","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:31:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"16f33a5f3f7f584b602e2fff4bd8891a6abb900f","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:58:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75dc887f6d9e40608066f14e2f381bac\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/77010af4f1cd4017ac3832dc16767867 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6beca8328bc474f73ba32096dc1f449a6c372832","revisions":{"6beca8328bc474f73ba32096dc1f449a6c372832":{"kind":"REWORK","_number":1,"created":"2022-10-05 15:12:52.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/74/860474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/74/860474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/860474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/860474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/74/860474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/74/860474/1"}}},"commit":{"parents":[{"commit":"41686c0bee0dd787c785b8c43c3a07cebb97d991","subject":"Merge \"charms: add new ceph subteam\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/41686c0bee0dd787c785b8c43c3a07cebb97d991"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-10-05 15:09:11.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-10-05 15:09:11.000000000","tz":-420},"subject":"Update OpenDev tenants\u0027 default ansible to 6","message":"Update OpenDev tenants\u0027 default ansible to 6\n\nZuul is looking at dropping ansible 5 support. By updating our default\nansible version to 6 early we can provide valuable feedback to the Zuul\nproject. That said we expect this to function. The Zuul Zuul tenant is\nalready running under ansible 6, and I\u0027ve done some testing with\ndevstack [0] that appears to show those jobs running happily.\n\nThe openstack release is now out the door which makes this a good time\nto change the zuul default.\n\n[0] https://review.opendev.org/c/openstack/devstack/+/858436\n\nChange-Id: I61493e74d9955a444326d72e085f511a838f88dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6beca8328bc474f73ba32096dc1f449a6c372832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6beca8328bc474f73ba32096dc1f449a6c372832"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
