)]}'
{"id":"openstack%2Fproject-config~861035","triplet_id":"openstack%2Fproject-config~master~I4a4d9cb6e292bb693ae424be0928501fed81113f","project":"openstack/project-config","branch":"master","topic":"constraints-py310","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-10-12 11:38:40.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-10-12 11:38:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4a4d9cb6e292bb693ae424be0928501fed81113f","subject":"Switch the requirements-constraints job to py310","status":"MERGED","created":"2022-10-12 05:25:30.000000000","updated":"2022-10-12 11:38:40.000000000","submitted":"2022-10-12 11:38:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861035-constraints-py310","meta_rev_id":"b5244439349c40407c77e080c2c6cc725a0f1f5f","_number":861035,"virtual_id_number":861035,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 11:38:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-10-12 10:07:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-12 11:30:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-12 11:30:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-12 05:32:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 05:37:23.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2022-10-12 11:30:44.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"d80392f7c914c198cb3194f79573655eb9b9adfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-12 05:25:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"236bd60e63de61294fa9fda34a05e135b0e4f0bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 05:32:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604c1af0279440bca751c189b3c851f1\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ec573a6d2c804195a6dd62b9dbcd393c : SUCCESS in 2m 43s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/046d7e09442c43e0879b1d60974755db : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"891315e11f3322e55b6110faec4a02cc394ec510","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-10-12 05:37:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4bc22b37cffbd080e813cc12f8ec93d0c14ddd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-10-12 05:38:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d09d5e1fc8b3845fee675aa057101bc3dba36b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 05:56:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02243e594a454cf483f7f1ff13c38f34\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/20674f71ac3343f3bdef1bdbe9154409 : SUCCESS in 2m 44s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/15bbdf4a2c43424ba753ed0ba9dd2188 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"ca3c32aed2070cbf2a7569ae05f0b9952c9164be","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-10-12 10:07:03.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2c837380dd44a073fb37cabaff2db8cf2415c2d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-12 11:30:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nWith the other reviewers involved, I\u0027m fairly comfortable approving.","accounts_in_message":[],"_revision_number":2},{"id":"21136897e7f42d84da1bc38eb37cad645d5c3813","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 11:31:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4db7839d6ee6cbe29201253204d36e95876637c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 11:38:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4787f6b520e4e31b5babf11f8711471\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e688864a8bf641f3a1e7f331eba2ba78 : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b5244439349c40407c77e080c2c6cc725a0f1f5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 11:38:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"702037af47f00639a93566335029c159a5a33284","revisions":{"bf873189119b30f4dd2c5b72065e74c34255d843":{"kind":"REWORK","_number":1,"created":"2022-10-12 05:25:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/35/861035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/861035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/861035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/861035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/861035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/861035/1"}}},"commit":{"parents":[{"commit":"b70a078917bb035880cef8dc8c4d85b712946ad4","subject":"Merge \"Replace old Zed cycle key with 2023.1/Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b70a078917bb035880cef8dc8c4d85b712946ad4"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-12 05:22:14.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-12 05:22:14.000000000","tz":120},"subject":"Switch the requirements-constraints job to py310","message":"Switch the requirements-constraints job to py310\n\nSince the PTI switched the supported python versions for the current\ncycle from 3.8+3.9 to 3.9+3.10, we want to update the constraints\ngeneration accordingly. Since we don\u0027t have a stable platform where we\ncan run both py3.9 and py3.10 in parallel, generate constraints only for\npy3.10 for now.\n\nChange-Id: I4a4d9cb6e292bb693ae424be0928501fed81113f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf873189119b30f4dd2c5b72065e74c34255d843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf873189119b30f4dd2c5b72065e74c34255d843"}]},"branch":"refs/heads/master"},"702037af47f00639a93566335029c159a5a33284":{"kind":"REWORK","_number":2,"created":"2022-10-12 05:38:30.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/35/861035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/861035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/861035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/861035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/861035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/861035/2"}}},"commit":{"parents":[{"commit":"b70a078917bb035880cef8dc8c4d85b712946ad4","subject":"Merge \"Replace old Zed cycle key with 2023.1/Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b70a078917bb035880cef8dc8c4d85b712946ad4"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-10-12 05:22:14.000000000","tz":120},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-10-12 05:38:30.000000000","tz":0},"subject":"Switch the requirements-constraints job to py310","message":"Switch the requirements-constraints job to py310\n\nSince the PTI switched the supported python versions for the current\ncycle from 3.8+3.9 to 3.9+3.10, we want to update the constraints\ngeneration accordingly. Since we don\u0027t have a stable platform where we\ncan run both py3.9 and py3.10 in parallel, generate constraints only for\npy3.10 for now.\n\nChange-Id: I4a4d9cb6e292bb693ae424be0928501fed81113f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/702037af47f00639a93566335029c159a5a33284"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/702037af47f00639a93566335029c159a5a33284"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
