)]}'
{"id":"openstack%2Fproject-config~861046","triplet_id":"openstack%2Fproject-config~master~I69a9c9f388371a7a8ad1c7fe5cfc8f6c444d4f45","project":"openstack/project-config","branch":"master","topic":"add-charm-jobs","attention_set":{},"removed_from_attention_set":{"12549":{"account":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"last_update":"2022-10-12 16:21:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69a9c9f388371a7a8ad1c7fe5cfc8f6c444d4f45","subject":"Add project for managing zuul jobs for charms","status":"MERGED","created":"2022-10-12 08:51:49.000000000","updated":"2022-10-12 16:39:06.000000000","submitted":"2022-10-12 16:21:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"861046-add-charm-jobs","meta_rev_id":"57d1b6660ef0b1c2f81dfa27186d00f51f04f370","_number":861046,"virtual_id_number":861046,"owner":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 16:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"all":[{"value":1,"date":"2022-10-12 15:31:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-12 11:31:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-12 16:08:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-12 08:59:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 11:31:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-10-12 15:31:16.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"}],"messages":[{"id":"701b77d45bb8ab663e8fa2769778f6e4a4323ed4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2022-10-12 08:51:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebcd00af97a893bcae56dc81c2b5b23a4e933710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 08:59:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eab15a070ec4a5d8fd351238d323581\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/860fb0d7730e4c9a95161cb89edccc15 : SUCCESS in 2m 36s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/85704ab0f0334a2a874efda7e3e999ee : SUCCESS in 6m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/696ab4cc118c4742917e7c78b2168a57 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ebaa95db0818ba3f9248bc2907a3436e7a5d1713","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-12 11:31:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3aebf1d5f2da3a9b2abbafe96d88f0e60b656d2c","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2022-10-12 15:31:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1eb4e6e55c9e67b7089008d8c1d117b026c3814","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-10-12 16:08:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fad24598d5d92e2c1224af4a9e5760a22a816d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 16:09:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4fe5c91450f106792dcaf4edd0acbb0c336cd623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 16:21:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f140137429aa4592972cc1f247cc91a7\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/71695f7843b94980a1177d600ca2225c : SUCCESS in 7m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/20d732defa084201a0cd3914baad86d7 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9dddc4c3567a38393c1fdf3a88a2d10269e643db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 16:21:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"57d1b6660ef0b1c2f81dfa27186d00f51f04f370","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 16:39:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d01e39130f4741c598b57f4106f5af1e\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/ca5c4016b7a545a7a0b274793698c504 : SUCCESS in 7m 21s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/391b1e16a2674f4e94e283e9ed531089 : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a0412959385ee152318dab0e5495ec9e4f38d2f","revisions":{"6a0412959385ee152318dab0e5495ec9e4f38d2f":{"kind":"REWORK","_number":1,"created":"2022-10-12 08:51:49.000000000","uploader":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"ref":"refs/changes/46/861046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/46/861046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/861046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/861046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/46/861046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/46/861046/1"}}},"commit":{"parents":[{"commit":"b70a078917bb035880cef8dc8c4d85b712946ad4","subject":"Merge \"Replace old Zed cycle key with 2023.1/Antelope\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b70a078917bb035880cef8dc8c4d85b712946ad4"}]}],"author":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2022-10-12 08:39:34.000000000","tz":60},"committer":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2022-10-12 08:41:31.000000000","tz":60},"subject":"Add project for managing zuul jobs for charms","message":"Add project for managing zuul jobs for charms\n\nAdd a project for defining the zuul jobs for testing charms.\n\nChange-Id: I69a9c9f388371a7a8ad1c7fe5cfc8f6c444d4f45\nRequired-By: I07ace283a8255f106c141d7bea06618f802e3b51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a0412959385ee152318dab0e5495ec9e4f38d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6a0412959385ee152318dab0e5495ec9e4f38d2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
