)]}'
{"id":"openstack%2Fproject-config~863025","triplet_id":"openstack%2Fproject-config~master~I429fd355819b0bff72994ff1b531d36b0ce70831","project":"openstack/project-config","branch":"master","topic":"add-cfn-repo","attention_set":{},"removed_from_attention_set":{"34726":{"account":{"_account_id":34726,"name":"Jie Niu","username":"jieniu","inactive":true},"last_update":"2022-11-04 13:55:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I429fd355819b0bff72994ff1b531d36b0ce70831","subject":"Apply cfn repository for code and storyboard","status":"ABANDONED","created":"2022-10-31 09:42:47.000000000","updated":"2022-11-04 13:55:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac0b99b204aae1c52271ce9f3d62f32cb93c7073","_number":863025,"virtual_id_number":863025,"owner":{"_account_id":34726,"name":"Jie Niu","username":"jieniu","inactive":true},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-11-01 03:51:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-31 10:10:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc35620328f47363570a4ec45a6d4323849cc509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34726,"name":"Jie Niu","username":"jieniu","inactive":true},"date":"2022-10-31 09:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a9827597af58001a06e18d7f4ae8aa5cb6bd03d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 10:10:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/120d9db8f4bf4f459eae46644ce05294\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/160e39a218eb4f60b65f24215b17a19a : SUCCESS in 3m 06s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c2301cac5ea54a138644f95b10e3468c : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4681d844142f4b8e9cda8ac7506bde35 : FAILURE in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"560f17765d8a238c6198bbe25536613766f0837c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34726,"name":"Jie Niu","username":"jieniu","inactive":true},"date":"2022-11-01 03:38:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bced71e68c9d4370952b76c652279dcea677ccd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 03:51:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80f705d3866b444a919312ebfa9c6b3a\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f60d7925b1594ee6a40092013e09d00b : SUCCESS in 3m 31s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5dc3264f4ccc4e088f4e2184592cb54b : SUCCESS in 6m 13s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2c9399daef674a38bd225071084aa88d : FAILURE in 3m 23s","accounts_in_message":[],"_revision_number":2},{"id":"ac0b99b204aae1c52271ce9f3d62f32cb93c7073","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-11-04 13:55:07.000000000","message":"Abandoned\n\nAbandon as this is continued in https://review.opendev.org/c/openstack/project-config/+/863168","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cb410e8030f6fece679dd6db5b7d841dec01853e","revisions":{"bf18ca808f2f7598c77fc11f0a3fde99dc1e45f9":{"kind":"REWORK","_number":1,"created":"2022-10-31 09:42:47.000000000","uploader":{"_account_id":34726,"name":"Jie Niu","username":"jieniu","inactive":true},"ref":"refs/changes/25/863025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/863025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/863025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/863025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/863025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/863025/1"}}},"commit":{"parents":[{"commit":"a4ca89d746b690ec15b5014d9fa7e9b725b51147","subject":"Merge \"Mirror keystone-openidc to github\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4ca89d746b690ec15b5014d9fa7e9b725b51147"}]}],"author":{"name":"jieniucmcc","email":"niujie@chinamobile.com","date":"2022-10-31 09:41:11.000000000","tz":480},"committer":{"name":"jieniucmcc","email":"niujie@chinamobile.com","date":"2022-10-31 09:41:11.000000000","tz":480},"subject":"Apply cfn repository for code and storyboard","message":"Apply cfn repository for code and storyboard\n\nChange-Id: I429fd355819b0bff72994ff1b531d36b0ce70831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf18ca808f2f7598c77fc11f0a3fde99dc1e45f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bf18ca808f2f7598c77fc11f0a3fde99dc1e45f9"}]},"branch":"refs/heads/master"},"cb410e8030f6fece679dd6db5b7d841dec01853e":{"kind":"REWORK","_number":2,"created":"2022-11-01 03:38:56.000000000","uploader":{"_account_id":34726,"name":"Jie Niu","username":"jieniu","inactive":true},"ref":"refs/changes/25/863025/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/25/863025/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/863025/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/863025/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/25/863025/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/25/863025/2"}}},"commit":{"parents":[{"commit":"a4ca89d746b690ec15b5014d9fa7e9b725b51147","subject":"Merge \"Mirror keystone-openidc to github\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4ca89d746b690ec15b5014d9fa7e9b725b51147"}]}],"author":{"name":"jieniucmcc","email":"niujie@chinamobile.com","date":"2022-10-31 09:41:11.000000000","tz":480},"committer":{"name":"jieniucmcc","email":"niujie@chinamobile.com","date":"2022-11-01 03:32:19.000000000","tz":480},"subject":"Apply cfn repository for code and storyboard","message":"Apply cfn repository for code and storyboard\n\nChange-Id: I429fd355819b0bff72994ff1b531d36b0ce70831\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb410e8030f6fece679dd6db5b7d841dec01853e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb410e8030f6fece679dd6db5b7d841dec01853e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
