)]}'
{"id":"openstack%2Fproject-config~863418","triplet_id":"openstack%2Fproject-config~master~I6eecf834fe7b357f9ba4126135487e09e51900e2","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"1":{"account":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"last_update":"2022-11-02 17:27:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6eecf834fe7b357f9ba4126135487e09e51900e2","subject":"Update ZK followers graph","status":"MERGED","created":"2022-11-02 16:49:36.000000000","updated":"2022-11-02 17:39:28.000000000","submitted":"2022-11-02 17:27:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863418","meta_rev_id":"1210db555592e4be69b17c9a17894315e89d52c7","_number":863418,"virtual_id_number":863418,"owner":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-02 17:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-11-02 16:50:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2022-11-02 16:50:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-02 16:50:02.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-11-02 17:11:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b332f35fb43886908deca7fb9c09459baa5fbbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2022-11-02 16:49:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f14b3ecfdb3ce3fecbaf5870931c7959ea4a29f8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-11-02 16:50:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"89ef50927bb5bf4a48b3c008cc426d3242086774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 17:11:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88104a683b194e33968447f5cc1e844c\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/3b07d1464e1440879c7fae9f6e10fb48 : SUCCESS in 19m 42s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/611c6085bd7e4010aaa66bba3f57b4f6 : SUCCESS in 6m 25s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/ada971ba42774d27a316db8f106ffb60 : SUCCESS in 12m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0de241764f632f6c5a5cda6916444abd10e514a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 17:11:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"670d1088d3a0a9748fae7780be2507a6cce3d0c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 17:27:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab13d38c90841329d8af24ffc8b1a0d\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b416aff996fa4257a5dfb940834b1577 : SUCCESS in 14m 45s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ee9776f8c710442cb57cab542406ee21 : SUCCESS in 8m 26s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/c4c869c2adc443709e65c1a52bffad59 : SUCCESS in 11m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e460a9b001c49c9ca2bd28115473e375844cff58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 17:27:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1210db555592e4be69b17c9a17894315e89d52c7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 17:39:28.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/667bc50b07a543c0a39c59599924f2f2\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/c9bcacc2da5f42d1bfb64c3e88786b34 : RETRY_LIMIT in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e15b9cae77bdc243322cee64b3688a2a43dd193c","revisions":{"e15b9cae77bdc243322cee64b3688a2a43dd193c":{"kind":"REWORK","_number":1,"created":"2022-11-02 16:49:36.000000000","uploader":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"ref":"refs/changes/18/863418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/863418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/863418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/863418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/863418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/863418/1"}}},"commit":{"parents":[{"commit":"a4ca89d746b690ec15b5014d9fa7e9b725b51147","subject":"Merge \"Mirror keystone-openidc to github\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a4ca89d746b690ec15b5014d9fa7e9b725b51147"}]}],"author":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2022-11-02 16:48:36.000000000","tz":-420},"committer":{"name":"James E. Blair","email":"jim@acmegating.com","date":"2022-11-02 16:48:36.000000000","tz":-420},"subject":"Update ZK followers graph","message":"Update ZK followers graph\n\nZK 3.6 no longer emits zk_followers so the best source for this\ngraph is zk_synced_followers now.\n\nChange-Id: I6eecf834fe7b357f9ba4126135487e09e51900e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
