)]}'
{"id":"openstack%2Fproject-config~863931","triplet_id":"openstack%2Fproject-config~master~I255a9c7070c268f48504f688a25e4de79a8b158b","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-12-06 14:45:47.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2022-12-06 14:45:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I255a9c7070c268f48504f688a25e4de79a8b158b","subject":"Allow Ironic/Sushy cores to toggle WIP state","status":"MERGED","created":"2022-11-07 20:42:39.000000000","updated":"2022-12-06 14:53:39.000000000","submitted":"2022-12-06 14:45:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863931","meta_rev_id":"6af172fd87913a24439bb9e2936b08875b3ae433","_number":863931,"virtual_id_number":863931,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-06 14:45:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2022-12-06 14:38:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-01 20:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2022-12-06 14:38:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 20:51:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-07 20:52:54.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-11-07 21:11:31.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2022-11-08 10:13:51.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2022-11-21 14:26:50.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-12-06 14:38:37.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"00683f4fa9141e571446b5046ac24ebc3c7fd995","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-11-07 20:42:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c15b2ebf7c51a215848ff84f212424a9c88e4807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:51:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c1ad7365e60451a8401b73fabb8ff4e\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/41128c8d2d5241099e371ed0659a1888 : SUCCESS in 5m 09s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9b2c599431654a64b6641561c0a46378 : SUCCESS in 6m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3644a58f0f064ab98aed38633750da33 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"3dfd56b3c3759d3fd2c39f6c438d86fbf9bf0156","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-07 20:52:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7ce5452580a87a18e5fcc2e2df579821d63d89c","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-11-07 20:53:40.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67843994f14c3054e7cce4d41448c14f37338dde","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-11-07 21:11:31.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e5d67e28b8ec1d116f6c6be11ebb64b4fad4650d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-11-07 21:15:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6f9087d06ce13e668016bb1f1eef130d687ca56b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 21:21:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/effa0ad27f29426eb1601b25bb6be69d\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ab45bdc493734275aae794d0e786df76 : SUCCESS in 2m 31s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c7e2cba74e5345b2b6a5277790e8d57a : SUCCESS in 4m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/821c408973d64bb2bdd2287f85f73da9 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"9ed0d5359f30bd436fc563a0b0ab939f56c05346","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-08 00:23:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"415192b943cbd764a081b3ad99618c07328f4b40","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-11-08 10:13:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1a984539d3b8fc7e031035e77a37c9560e7944b0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-08 15:39:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae5f6122d28ea1997d24498e1be50f8eee97e33b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-11-16 21:34:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c5768a9760b371a55cba052a2a471840cb0e186b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-11-16 21:34:30.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4197ddfe42e07e083166bcd42f0e26144236968c","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-11-16 21:34:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e58233c9bc4758ea46635890a724912099f3e592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 21:40:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d2dda90484b4c57849e702ff7771ed4\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/72be6165adee48a1962b1e28c3fac5fc : SUCCESS in 2m 34s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c1b99fc3f90749399000bda98ab49b5f : SUCCESS in 4m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9beafa46a3b64bfebf4aeb3776ba45a2 : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"628a9bec3ad07952e958a4ed802cdd26c2570e1c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-16 21:51:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"59be2497a42f454314594c0aa082de50b94db864","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-11-21 14:26:50.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7efc399a8bfa57edf45737b946077bcbbf5d392f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-12-01 19:54:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f17bdb82aa76340eb80ea62c45ee6cb06f734c24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 20:01:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fbd8d6a1b224b4f9fbc4f01dcc8ddd9\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/0688ae49c3094ae89da1f5a9f2646c27 : SUCCESS in 3m 54s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/45c221fce4364076b5c51278aeed162b : SUCCESS in 5m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d4ff2ac1f6bc42608a28c1bf0c88e011 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":4},{"id":"81d9d7582b74e7db37948f23554bc00eae162444","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-01 20:02:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cf780ba8d522331bd51ac22aea04d2921ad43374","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-12-06 14:38:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"75918e52db6a0085e3fdf263ca6e141b752d8f6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 14:39:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"43437e6f60afa8ec6628dc72613b9681a654bbec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 14:45:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d1b24f248c2446d8d35dbd809662246\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0126d94c7ec5417e8fb4b0979d443633 : SUCCESS in 6m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/56431f2bb6ee43c3bb62e4f9f25ab46a : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":4},{"id":"2612757100926976049a920f3c40eb1b9ed34123","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 14:45:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"6af172fd87913a24439bb9e2936b08875b3ae433","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 14:53:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0286ae6688a74fc6902183d07098cc68\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/793aa024a95143dabb08854ca7b0f0aa : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9cc2916a5415f3e2c3195a08b1efc9b507379c36","revisions":{"17e9d3804780e17aec17e92917f22cb36a0a9cf3":{"kind":"REWORK","_number":1,"created":"2022-11-07 20:42:39.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/31/863931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/863931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/863931/1"}}},"commit":{"parents":[{"commit":"e15b9cae77bdc243322cee64b3688a2a43dd193c","subject":"Update ZK followers graph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-11-07 20:40:22.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-11-07 20:42:37.000000000","tz":-480},"subject":"Allow Ironic cores to toggle WIP state","message":"Allow Ironic cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: I255a9c7070c268f48504f688a25e4de79a8b158b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/17e9d3804780e17aec17e92917f22cb36a0a9cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/17e9d3804780e17aec17e92917f22cb36a0a9cf3"}]},"branch":"refs/heads/master"},"dd0fcc04126409d093c7658e62c2abdb079d4731":{"kind":"REWORK","_number":2,"created":"2022-11-07 21:15:19.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/31/863931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/863931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/863931/2"}}},"commit":{"parents":[{"commit":"e15b9cae77bdc243322cee64b3688a2a43dd193c","subject":"Update ZK followers graph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-11-07 20:40:22.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-11-07 21:15:13.000000000","tz":-480},"subject":"Allow Ironic cores to toggle WIP state","message":"Allow Ironic cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: I255a9c7070c268f48504f688a25e4de79a8b158b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd0fcc04126409d093c7658e62c2abdb079d4731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dd0fcc04126409d093c7658e62c2abdb079d4731"}]},"branch":"refs/heads/master"},"dc6b5ace5a7d19a18d996230a04efed2f9d9dfd1":{"kind":"REWORK","_number":3,"created":"2022-11-16 21:34:27.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/31/863931/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/863931/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/863931/3"}}},"commit":{"parents":[{"commit":"e15b9cae77bdc243322cee64b3688a2a43dd193c","subject":"Update ZK followers graph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-11-07 20:40:22.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-11-16 21:34:21.000000000","tz":-480},"subject":"Allow Ironic cores to toggle WIP state","message":"Allow Ironic cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: I255a9c7070c268f48504f688a25e4de79a8b158b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dc6b5ace5a7d19a18d996230a04efed2f9d9dfd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dc6b5ace5a7d19a18d996230a04efed2f9d9dfd1"}]},"branch":"refs/heads/master"},"9cc2916a5415f3e2c3195a08b1efc9b507379c36":{"kind":"REWORK","_number":4,"created":"2022-12-01 19:54:08.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/31/863931/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/31/863931/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/31/863931/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/31/863931/4"}}},"commit":{"parents":[{"commit":"e15b9cae77bdc243322cee64b3688a2a43dd193c","subject":"Update ZK followers graph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-11-07 20:40:22.000000000","tz":-480},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-12-01 19:53:49.000000000","tz":-480},"subject":"Allow Ironic/Sushy cores to toggle WIP state","message":"Allow Ironic/Sushy cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: I255a9c7070c268f48504f688a25e4de79a8b158b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9cc2916a5415f3e2c3195a08b1efc9b507379c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9cc2916a5415f3e2c3195a08b1efc9b507379c36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}}]}],"submit_requirements":[]}
