)]}'
{"id":"openstack%2Fproject-config~863932","triplet_id":"openstack%2Fproject-config~master~I88e5e0df93b20ebf0db7e10462af051526541780","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2022-12-09 13:38:10.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-11-08 00:20:42.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I88e5e0df93b20ebf0db7e10462af051526541780","subject":"Allow Designate cores to toggle WIP state","status":"MERGED","created":"2022-11-07 20:50:18.000000000","updated":"2022-12-09 13:47:05.000000000","submitted":"2022-12-09 13:38:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863932","meta_rev_id":"af85f7b7ca5b23b2f8f05949ba09c81b34e98926","_number":863932,"virtual_id_number":863932,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-09 13:38:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-08 00:20:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-09 13:27:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-07 20:56:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-07 20:59:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2022-11-08 00:20:42.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"b2ff68a457456432b3c08178525ec506c6a9eda5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-07 20:50:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73d77db9803a7e24c22ec6c74aff301015ca82c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 20:56:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae756afc58e94668937f335017770e06\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/1242cb6938f3466cb834e4bf7d747816 : SUCCESS in 2m 48s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ccec8c8aa7744960ab394d001189ba4a : SUCCESS in 5m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8dbf110447344c88b5cee5ef2819f43f : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"5cc13bffd1b47ad4dd56d58010b76af6a92b3a15","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-07 20:59:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9870c021796c6e9f49ca776df836b8326ddc9f0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-07 21:18:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2836a71c7a863c0560c4bebdc6df1ad167f1e542","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-07 21:20:49.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"dba0ab52b28c2438e02c486d7f9f7a0c0f7343d3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-07 21:20:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6f4e8a779e5c2bfea8251fdffa3b9c1e748cd57a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2022-11-07 21:24:51.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7a1a44c7d4147b88b6463d5ca371be771206a4b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 21:31:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe13eaf0f5b4570922a097b2e71f8a6\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/f4d3fe85f4ac4c9ba1e5b7431a6c385c : SUCCESS in 2m 35s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/852b716688c54502901f93eb476c6be8 : SUCCESS in 4m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/01391929df1540d8bf71746fcfcedabb : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":4},{"id":"114c1bb3cd87098805775e9233b2f19cf3e47d2a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-08 00:20:42.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"90924f52bbdcbabbee5dcd437fb9afcacf6c5006","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-09 13:27:27.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c5610b616de74e2f258d985f6cf91b591a596097","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 13:27:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"19d380a7da205e3eac72d8bd072d8189105dddee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 13:38:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cebfbaaacd04dab9e3faca30355b0c2\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f2dbc152a69c47e893da651ea38f5415 : SUCCESS in 5m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3f11010236fb49789cd27e62a4337fde : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":4},{"id":"c9b48a85f12a07108326b792928b39ae39656178","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 13:38:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"af85f7b7ca5b23b2f8f05949ba09c81b34e98926","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 13:47:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c815dcbdc3e442e1a4f69522976359a9\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/adccf273698a460c9ac0168988153b4e : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1d500fa9a99402afb1ee99b55e0616882ee6c250","revisions":{"2255f9265b45b7d8d4fe53f015cd91f178abd253":{"kind":"REWORK","_number":1,"created":"2022-11-07 20:50:18.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/863932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/863932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/863932/1"}}},"commit":{"parents":[{"commit":"e15b9cae77bdc243322cee64b3688a2a43dd193c","subject":"Update ZK followers graph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-07 20:49:28.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-07 20:50:16.000000000","tz":0},"subject":"Allow Designate cores to toggle WIP state","message":"Allow Designate cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: I88e5e0df93b20ebf0db7e10462af051526541780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2255f9265b45b7d8d4fe53f015cd91f178abd253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2255f9265b45b7d8d4fe53f015cd91f178abd253"}]},"branch":"refs/heads/master"},"0f5c0375fe69f896cd5926db46d5c92d795d50e2":{"kind":"REWORK","_number":2,"created":"2022-11-07 21:18:35.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/863932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/863932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/863932/2"}}},"commit":{"parents":[{"commit":"e15b9cae77bdc243322cee64b3688a2a43dd193c","subject":"Update ZK followers graph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-07 20:49:28.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-07 21:18:34.000000000","tz":0},"subject":"Allow Designate cores to toggle WIP state","message":"Allow Designate cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: I88e5e0df93b20ebf0db7e10462af051526541780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f5c0375fe69f896cd5926db46d5c92d795d50e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f5c0375fe69f896cd5926db46d5c92d795d50e2"}]},"branch":"refs/heads/master"},"e64829b9ac031ca235477f99422e66cf931535f7":{"kind":"REWORK","_number":3,"created":"2022-11-07 21:20:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/863932/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/863932/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/863932/3"}}},"commit":{"parents":[{"commit":"e15b9cae77bdc243322cee64b3688a2a43dd193c","subject":"Update ZK followers graph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-07 20:49:28.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-07 21:19:31.000000000","tz":0},"subject":"Allow Designate cores to toggle WIP state","message":"Allow Designate cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: I88e5e0df93b20ebf0db7e10462af051526541780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e64829b9ac031ca235477f99422e66cf931535f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e64829b9ac031ca235477f99422e66cf931535f7"}]},"branch":"refs/heads/master"},"1d500fa9a99402afb1ee99b55e0616882ee6c250":{"kind":"REWORK","_number":4,"created":"2022-11-07 21:24:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/863932/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/32/863932/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/32/863932/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/32/863932/4"}}},"commit":{"parents":[{"commit":"e15b9cae77bdc243322cee64b3688a2a43dd193c","subject":"Update ZK followers graph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e15b9cae77bdc243322cee64b3688a2a43dd193c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-07 20:49:28.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2022-11-07 21:24:49.000000000","tz":0},"subject":"Allow Designate cores to toggle WIP state","message":"Allow Designate cores to toggle WIP state\n\nWIP state in Gerrit is tricky and has side effects a lot of new\ncontributors might not expect it to completely remove their patch\nfrom the dashboards. In those cases; it\u0027s nice for a core to be\nable to toggle the wip state.\n\nChange-Id: I88e5e0df93b20ebf0db7e10462af051526541780\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1d500fa9a99402afb1ee99b55e0616882ee6c250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1d500fa9a99402afb1ee99b55e0616882ee6c250"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
