)]}'
{"id":"openstack%2Fproject-config~863955","triplet_id":"openstack%2Fproject-config~master~I51107d5483bf2326941bac720b495cdf3cfb0e8d","project":"openstack/project-config","branch":"master","topic":"charm-infinidat-tools","attention_set":{},"removed_from_attention_set":{"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2022-11-11 07:55:40.000000000","reason":"\u003cGERRIT_ACCOUNT_6547\u003e replied on the change","reason_account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},"30110":{"account":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"last_update":"2022-11-30 20:16:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I51107d5483bf2326941bac720b495cdf3cfb0e8d","subject":"Add the infinidat-tools charm to Openstack charms","status":"MERGED","created":"2022-11-08 01:52:26.000000000","updated":"2022-11-30 22:58:35.000000000","submitted":"2022-11-30 20:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863955-charm-infinidat-tools","meta_rev_id":"786e0d684403aaa0ee4ef60af53738416053afef","_number":863955,"virtual_id_number":863955,"owner":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-30 20:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2022-11-11 07:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2022-11-21 08:09:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-14 20:14:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2022-11-30 20:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2022-11-30 20:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 02:05:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-10 07:45:17.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2022-11-14 20:14:20.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-11-21 08:09:02.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-11-30 20:07:56.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"802f3ad2ed765421a2aaa7c44410fb2c1b351fa8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-08 01:52:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a1337c55905397372b973654b5074cfd3e8cb2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 02:05:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a40485d25df4cbdb733bf253b965d46\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/61bd5316f75e418a9df18da7e25b1626 : SUCCESS in 2m 48s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dee7a8045b5244889b3502b8ebf4a59f : SUCCESS in 6m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f1b8874e668b4827888749e51f5466bb : FAILURE in 3m 09s","accounts_in_message":[],"_revision_number":1},{"id":"80d6409b66d8b6f4d240f1b4057ee3fa4b7a542d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-08 02:45:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5ed4038b193d360d64a43b635876e840af3a619b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-08 02:48:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e878b653d6014d3b140d59787fd5d01093b80100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 02:57:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df38ff69c543493ba2499a88abd5454a\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7539f51e31ea47c89bb0c77dc3b4b8df : SUCCESS in 6m 25s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8d06e7984461481c94e1f374c27a394b : SUCCESS in 5m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/38ac2a5f5a564c4793f1920786fe6257 : FAILURE in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"7f58a00d12e792f6fe47fbe3f656731ba997faac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-08 21:58:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d95233f8ffb85e900edf42621069463d186f2642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 22:07:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee1a4195766a491dbeedd8bb3d68f377\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/fcf7a2b2bc6d4ae5ad514a652416c12e : SUCCESS in 4m 13s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7a2207256ec3424ba10af3cd737bd35c : SUCCESS in 7m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a110cc9bcee14e10983d7bba02342f21 : FAILURE in 4m 20s","accounts_in_message":[],"_revision_number":4},{"id":"d2cb5f97b5ec8f8e911604eef1100958d6207b3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-09 23:31:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"271e562b8d0389ea022776dea2b832acf4efb343","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 23:37:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/130131e4269b495fb960a376f9d1bd88\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ec2ce9130e6741d48f6f8096a44b6bcc : SUCCESS in 2m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0dceec9225de41ce8f9659755ed479a9 : SUCCESS in 5m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8db31ad792b2497da626de19e4750c0d : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":5},{"id":"258dde819853667c41224f82793adfd70cf3d40c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-11-10 07:45:17.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a9f47dc2acfbf2e31f6b9220b208bcd64750133a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-10 10:06:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c8d15bd985277ae30c2602076e6aa163547d3715","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 10:18:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61b7f21e97be4b35892f81019002826b\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/13b0965611f74487bbcfbbde22e2a4ff : SUCCESS in 2m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dd667f951ddd4c5180340d4f02d1f4ef : SUCCESS in 6m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/92bfe71435ce43ce9ab5258f89aac6e2 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":6},{"id":"6f24efd826e4c53ab2392f7cc3ca9b1d17cf3fbb","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-10 22:33:12.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7bd1d73c4a75ec482109d5f3fab11361d186b133","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-11-11 07:55:40.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b23e2b62fc76b5298b8e42b94a5a3ae54c80c88e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-14 20:14:20.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"0daa6ec77987b8112bc43e9e5af0259d72b9864f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-21 08:09:02.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"804ef76f311cb9f56e8df40fe883f9eb9cc396cd","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-30 20:07:56.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5b065bdc4b2790d0f664571ab6688df3c5855bb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 20:08:22.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7ef649519531128c81282f220bc2e3ca30aef356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 20:16:37.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f66cc542b7f1403fafe979ba25c8a4ad\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d22e9d4f209c41b5be1f918a6f733148 : SUCCESS in 7m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b0f05b037d0d440a982035549e4e6489 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":6},{"id":"a4b04049c43f97fdbc0d8acb47ca35c4677b786c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 20:16:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"786e0d684403aaa0ee4ef60af53738416053afef","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 22:58:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5afa22cc1b6843a3a461840e02ffc9c0\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/7734ec59749d487e9d346fd4d0ac5c5f : SUCCESS in 8m 08s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/a8e0b8fb16924e13b06655a1777522f1 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"963110054d923a334819f30b87c0f2eb69f236e2","revisions":{"bb360ebbf7921a6f36c4d86d17910b68b1cf7dc5":{"kind":"REWORK","_number":1,"created":"2022-11-08 01:52:26.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/55/863955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/863955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/863955/1"}}},"commit":{"parents":[{"commit":"047f82a5fc69762872e417764f8a19c8603bb8ca","subject":"Add the cinder-infinidat charm to Openstack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/047f82a5fc69762872e417764f8a19c8603bb8ca"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:50:54.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:50:54.000000000","tz":180},"subject":"Add the infinidat-tools charm to Openstack charms","message":"Add the infinidat-tools charm to Openstack charms\n\nChange-Id: I51107d5483bf2326941bac720b495cdf3cfb0e8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb360ebbf7921a6f36c4d86d17910b68b1cf7dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bb360ebbf7921a6f36c4d86d17910b68b1cf7dc5"}]},"branch":"refs/heads/master"},"abbc536990bfebbe8fa421fea75b1e4558650092":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-11-08 02:45:16.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/55/863955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/863955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/863955/2"}}},"commit":{"parents":[{"commit":"f6112eb9482efe75efa7a05dd62dc07adcb638ce","subject":"Add the cinder-infinidat charm to Openstack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6112eb9482efe75efa7a05dd62dc07adcb638ce"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:50:54.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 02:44:35.000000000","tz":180},"subject":"Add the infinidat-tools charm to Openstack charms","message":"Add the infinidat-tools charm to Openstack charms\n\nChange-Id: I51107d5483bf2326941bac720b495cdf3cfb0e8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abbc536990bfebbe8fa421fea75b1e4558650092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abbc536990bfebbe8fa421fea75b1e4558650092"}]},"branch":"refs/heads/master"},"24386aefa03b346a582f2776968b2782f3bffcfa":{"kind":"REWORK","_number":3,"created":"2022-11-08 02:48:22.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/55/863955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/863955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/863955/3"}}},"commit":{"parents":[{"commit":"f6112eb9482efe75efa7a05dd62dc07adcb638ce","subject":"Add the cinder-infinidat charm to Openstack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f6112eb9482efe75efa7a05dd62dc07adcb638ce"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:50:54.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 02:46:33.000000000","tz":180},"subject":"Add the infinidat-tools charm to Openstack charms","message":"Add the infinidat-tools charm to Openstack charms\n\nChange-Id: I51107d5483bf2326941bac720b495cdf3cfb0e8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24386aefa03b346a582f2776968b2782f3bffcfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24386aefa03b346a582f2776968b2782f3bffcfa"}]},"branch":"refs/heads/master"},"8f35113bb2b10918ee73d257f49bd67fc8fe0fac":{"kind":"REWORK","_number":4,"created":"2022-11-08 21:58:54.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/55/863955/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/863955/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/863955/4"}}},"commit":{"parents":[{"commit":"1ea6374489e2cb4c7e776b8110f99f8d1d5ff766","subject":"Skip existing remote artifacts during PyPI upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ea6374489e2cb4c7e776b8110f99f8d1d5ff766"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:50:54.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 21:56:42.000000000","tz":180},"subject":"Add the infinidat-tools charm to Openstack charms","message":"Add the infinidat-tools charm to Openstack charms\n\nChange-Id: I51107d5483bf2326941bac720b495cdf3cfb0e8d\nNeeded-By: I081928fd0e789cca793ce1575127ceec46c906be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f35113bb2b10918ee73d257f49bd67fc8fe0fac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8f35113bb2b10918ee73d257f49bd67fc8fe0fac"}]},"branch":"refs/heads/master"},"fcda598acfdfd98b130867d6c07e08be6197f530":{"kind":"REWORK","_number":5,"created":"2022-11-09 23:31:29.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/55/863955/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/863955/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/863955/5"}}},"commit":{"parents":[{"commit":"1ea6374489e2cb4c7e776b8110f99f8d1d5ff766","subject":"Skip existing remote artifacts during PyPI upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ea6374489e2cb4c7e776b8110f99f8d1d5ff766"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:50:54.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-09 23:31:05.000000000","tz":180},"subject":"Add the infinidat-tools charm to Openstack charms","message":"Add the infinidat-tools charm to Openstack charms\n\nChange-Id: I51107d5483bf2326941bac720b495cdf3cfb0e8d\nNeeded-By: I081928fd0e789cca793ce1575127ceec46c906be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcda598acfdfd98b130867d6c07e08be6197f530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcda598acfdfd98b130867d6c07e08be6197f530"}]},"branch":"refs/heads/master"},"963110054d923a334819f30b87c0f2eb69f236e2":{"kind":"REWORK","_number":6,"created":"2022-11-10 10:06:44.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/55/863955/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/863955/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/863955/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/863955/6"}}},"commit":{"parents":[{"commit":"1ea6374489e2cb4c7e776b8110f99f8d1d5ff766","subject":"Skip existing remote artifacts during PyPI upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ea6374489e2cb4c7e776b8110f99f8d1d5ff766"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:50:54.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-10 10:06:22.000000000","tz":180},"subject":"Add the infinidat-tools charm to Openstack charms","message":"Add the infinidat-tools charm to Openstack charms\n\nChange-Id: I51107d5483bf2326941bac720b495cdf3cfb0e8d\nNeeded-By: I081928fd0e789cca793ce1575127ceec46c906be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/963110054d923a334819f30b87c0f2eb69f236e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/963110054d923a334819f30b87c0f2eb69f236e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
