)]}'
{"id":"openstack%2Fproject-config~863957","triplet_id":"openstack%2Fproject-config~master~Iff96fbd4d01fd8eb2bbcfb7080244d7a40811434","project":"openstack/project-config","branch":"master","topic":"charm-manila-infinidat","attention_set":{},"removed_from_attention_set":{"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2022-11-11 07:55:26.000000000","reason":"\u003cGERRIT_ACCOUNT_6547\u003e replied on the change","reason_account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},"30110":{"account":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"last_update":"2022-11-30 20:16:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iff96fbd4d01fd8eb2bbcfb7080244d7a40811434","subject":"Add manila-infinidat charm to OpenStack charms","status":"MERGED","created":"2022-11-08 02:45:16.000000000","updated":"2022-11-30 23:16:17.000000000","submitted":"2022-11-30 20:16:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"863957-charm-manila-infinidat","meta_rev_id":"4512365bf1be6ed53a70a1183a56a1be19aa4553","_number":863957,"virtual_id_number":863957,"owner":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-30 20:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2022-11-11 07:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2022-11-21 08:08:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-11-14 20:14:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-30 20:08:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-30 20:08:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 03:00:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-10 07:45:55.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2022-11-14 20:14:39.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-11-21 08:08:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-11-30 20:08:14.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"15bb3cde6495f807dfd72d4471633720ae489ff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-08 02:45:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16bf98854de55ca9a523bb916ebeac748761b092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-08 02:48:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08b56d3c5f6776a4ce416695cef469807f1c9dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 03:00:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26125c2272354e31a80db787bef0873f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8fc798d8bee54eeeabd2917a6c919209 : SUCCESS in 2m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/48fd5d5c95a844a986f3fd80c41e75a1 : SUCCESS in 6m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5c38096ebf5749c5be8ca76c15da5378 : FAILURE in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"74718ca90ab85817523ee21ff96d9cd799340c48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-08 22:51:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f6fbc8a52ef390531d8f97c3e090ece4ebed0b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 23:03:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/249ffde1eb5f4c849815bd370c7055da\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/405e92d886ff460b8371ac9fb5fa90ef : SUCCESS in 2m 55s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3d3718a290a34866a90792acba33aa25 : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c91816ea9dc74ce6861b279fa1d19110 : FAILURE in 3m 16s","accounts_in_message":[],"_revision_number":3},{"id":"5106c827a87292b45bbad7746d3879755fa55378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-09 23:33:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f54f5c26f48c3a733fd1e3224ba844a42fe0aa3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 23:43:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eaf1f9360b742acaf2daed9089c7f0d\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6de288093e684c4bb29ef988a27c13ab : SUCCESS in 2m 41s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cdbee2e1e3d44249b6c4a51c32e0cdfa : SUCCESS in 5m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d7ca500a7e149c5ac5691ac95c193fd : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":4},{"id":"8dab548acac296811dd7aacdbf831f2b627ec533","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-11-10 07:45:55.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bb9aab4686bbe740fed551d65d21b4a8c18d314d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-10 10:03:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4f7ac96692b35feedbbe7d678e305693420d505a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-10 10:08:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"928dd9ebf3705f3420e9f0bc35fba5d95eb3ab82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-10 10:09:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bc1f82c9110602f8dc56b78de1739b7bc2be8f82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 10:20:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/300dfc0ca9f2414386d3c41b84d6a853\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/6aa955f161eb48cdbc423a978e5face7 : SUCCESS in 2m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/efa5786f3a3341a6ad132332c9337a59 : SUCCESS in 6m 07s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a26653bf2cb243918f37dedf9eeff525 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":7},{"id":"d57feafcff91dd12faef7d807c47c72730322b6c","author":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"date":"2022-11-10 22:33:24.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e91302d91f34ce4e7b8db2dd8db0664a6b2f4d4d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-11-11 07:55:26.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c0e6314014492d675bd3c783559460ce6b4152c0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-14 20:14:39.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"0f4b47d615fbdfb90aa204a8b7effa41955db61b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-11-21 08:08:57.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"c03db33908644e995909e4883b63b534fc3e92fe","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-30 20:08:14.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"80765962f228491f0cd87a6042983a05e4a7e274","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 20:08:41.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"da6835927c71136a119f6002666585642c9bb74c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 20:16:39.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/265092c5d96044148d03c85be903f008\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/aedf6e7f298c4fba9903fc8ba42ca1f6 : SUCCESS in 5m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/886f2f8af18f431192e4ee43ed2640e4 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":7},{"id":"4cb238ea89c454221727644ca72f30a05cad2f74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 20:16:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"4512365bf1be6ed53a70a1183a56a1be19aa4553","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 23:16:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4fedb47c35f448cbda7b59faf6f894b\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/1453a9bec5fd4d548070b2435f55387b : SUCCESS in 8m 34s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/e8a77a03ad414d07ba471659095bc8ef : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"36ec25d96c263536ed105fc409669d009c7c8a07","revisions":{"4621266d25770888ace0691da6793f8d97ef3910":{"kind":"REWORK","_number":1,"created":"2022-11-08 02:45:16.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/57/863957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/863957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/863957/1"}}},"commit":{"parents":[{"commit":"abbc536990bfebbe8fa421fea75b1e4558650092","subject":"Add the infinidat-tools charm to Openstack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/abbc536990bfebbe8fa421fea75b1e4558650092"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:55:31.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 02:44:42.000000000","tz":180},"subject":"Add manila-infinidat backend charm to OpenStack charms","message":"Add manila-infinidat backend charm to OpenStack charms\n\nAdd new charm that provides the Infinidat backend support\nto the OpenStack Manila service.\n\nChange-Id: Iff96fbd4d01fd8eb2bbcfb7080244d7a40811434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4621266d25770888ace0691da6793f8d97ef3910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4621266d25770888ace0691da6793f8d97ef3910"}]},"branch":"refs/heads/master"},"555a9587b49774116502fffed68847770071c982":{"kind":"REWORK","_number":2,"created":"2022-11-08 02:48:22.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/57/863957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/863957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/863957/2"}}},"commit":{"parents":[{"commit":"24386aefa03b346a582f2776968b2782f3bffcfa","subject":"Add the infinidat-tools charm to Openstack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24386aefa03b346a582f2776968b2782f3bffcfa"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:55:31.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 02:47:33.000000000","tz":180},"subject":"Add manila-infinidat backend charm to OpenStack charms","message":"Add manila-infinidat backend charm to OpenStack charms\n\nAdd new charm that provides the Infinidat backend support\nto the OpenStack Manila service.\n\nChange-Id: Iff96fbd4d01fd8eb2bbcfb7080244d7a40811434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/555a9587b49774116502fffed68847770071c982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/555a9587b49774116502fffed68847770071c982"}]},"branch":"refs/heads/master"},"7694a7162ea17befeb618e56478e0c9ea2a25d3f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2022-11-08 22:51:40.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/57/863957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/863957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/863957/3"}}},"commit":{"parents":[{"commit":"1ea6374489e2cb4c7e776b8110f99f8d1d5ff766","subject":"Skip existing remote artifacts during PyPI upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ea6374489e2cb4c7e776b8110f99f8d1d5ff766"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:55:31.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 22:48:34.000000000","tz":180},"subject":"Add manila-infinidat backend charm to OpenStack charms","message":"Add manila-infinidat backend charm to OpenStack charms\n\nAdd new charm that provides the Infinidat backend support\nto the OpenStack Manila service.\n\nNeeded-by: I932b805e0f7d44040bad9c28ca4799f243f651af\nChange-Id: Iff96fbd4d01fd8eb2bbcfb7080244d7a40811434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7694a7162ea17befeb618e56478e0c9ea2a25d3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7694a7162ea17befeb618e56478e0c9ea2a25d3f"}]},"branch":"refs/heads/master"},"c616649d69e5d85b17112a4c0d2a93accb9b5c15":{"kind":"REWORK","_number":4,"created":"2022-11-09 23:33:36.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/57/863957/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/863957/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/863957/4"}}},"commit":{"parents":[{"commit":"1ea6374489e2cb4c7e776b8110f99f8d1d5ff766","subject":"Skip existing remote artifacts during PyPI upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ea6374489e2cb4c7e776b8110f99f8d1d5ff766"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:55:31.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-09 23:33:17.000000000","tz":180},"subject":"Add manila-infinidat backend charm to OpenStack charms","message":"Add manila-infinidat backend charm to OpenStack charms\n\nAdd new charm that provides the Infinidat backend support\nto the OpenStack Manila service.\n\nNeeded-by: I932b805e0f7d44040bad9c28ca4799f243f651af\nChange-Id: Iff96fbd4d01fd8eb2bbcfb7080244d7a40811434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c616649d69e5d85b17112a4c0d2a93accb9b5c15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c616649d69e5d85b17112a4c0d2a93accb9b5c15"}]},"branch":"refs/heads/master"},"e38e448b5e9c4d4ad5ee478a94b7cfb058512b5a":{"kind":"REWORK","_number":5,"created":"2022-11-10 10:03:25.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/57/863957/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/863957/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/863957/5"}}},"commit":{"parents":[{"commit":"1ea6374489e2cb4c7e776b8110f99f8d1d5ff766","subject":"Skip existing remote artifacts during PyPI upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ea6374489e2cb4c7e776b8110f99f8d1d5ff766"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:55:31.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-10 10:03:02.000000000","tz":180},"subject":"Add manila-infinidat backend charm to OpenStack charms","message":"Add manila-infinidat backend charm to OpenStack charms\n\nAdd new charm that provides the Infinidat backend support\nto the OpenStack Manila service.\n\nNeeded-by: I932b805e0f7d44040bad9c28ca4799f243f651af\nChange-Id: Iff96fbd4d01fd8eb2bbcfb7080244d7a40811434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e38e448b5e9c4d4ad5ee478a94b7cfb058512b5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e38e448b5e9c4d4ad5ee478a94b7cfb058512b5a"}]},"branch":"refs/heads/master"},"ae2205f185eba278dda9c874749af138ec344741":{"kind":"REWORK","_number":6,"created":"2022-11-10 10:08:45.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/57/863957/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/863957/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/863957/6"}}},"commit":{"parents":[{"commit":"1ea6374489e2cb4c7e776b8110f99f8d1d5ff766","subject":"Skip existing remote artifacts during PyPI upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ea6374489e2cb4c7e776b8110f99f8d1d5ff766"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:55:31.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-10 10:08:33.000000000","tz":180},"subject":"Add manila-infinidat charm to OpenStack charms","message":"Add manila-infinidat charm to OpenStack charms\n\nAdd new charm that provides the Infinidat backend support\nto the OpenStack Manila service.\n\nNeeded-by: I932b805e0f7d44040bad9c28ca4799f243f651af\nChange-Id: Iff96fbd4d01fd8eb2bbcfb7080244d7a40811434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae2205f185eba278dda9c874749af138ec344741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae2205f185eba278dda9c874749af138ec344741"}]},"branch":"refs/heads/master"},"36ec25d96c263536ed105fc409669d009c7c8a07":{"kind":"REWORK","_number":7,"created":"2022-11-10 10:09:25.000000000","uploader":{"_account_id":30110,"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","username":"nikolay.vinogradov"},"ref":"refs/changes/57/863957/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/863957/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/863957/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/863957/7"}}},"commit":{"parents":[{"commit":"1ea6374489e2cb4c7e776b8110f99f8d1d5ff766","subject":"Skip existing remote artifacts during PyPI upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1ea6374489e2cb4c7e776b8110f99f8d1d5ff766"}]}],"author":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-08 01:55:31.000000000","tz":180},"committer":{"name":"Nikolay Vinogradov","email":"nikolay.vinogradov@canonical.com","date":"2022-11-10 10:08:59.000000000","tz":180},"subject":"Add manila-infinidat charm to OpenStack charms","message":"Add manila-infinidat charm to OpenStack charms\n\nAdd new charm that provides the Infinidat backend support\nto the OpenStack Manila service.\n\nNeeded-by: I932b805e0f7d44040bad9c28ca4799f243f651af\nChange-Id: Iff96fbd4d01fd8eb2bbcfb7080244d7a40811434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36ec25d96c263536ed105fc409669d009c7c8a07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/36ec25d96c263536ed105fc409669d009c7c8a07"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
