)]}'
{"id":"openstack%2Fproject-config~865433","triplet_id":"openstack%2Fproject-config~master~I92bc12b8f712e28962d24dd6474cfce22b81222c","project":"openstack/project-config","branch":"master","topic":"unbound/networkmanager","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2022-11-29 03:18:59.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2022-11-29 13:19:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I92bc12b8f712e28962d24dd6474cfce22b81222c","subject":"Ensure NetworkManager doesn\u0027t override /etc/resolv.conf","status":"MERGED","created":"2022-11-23 14:12:54.000000000","updated":"2022-11-29 13:44:53.000000000","submitted":"2022-11-29 13:19:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865433-unbound/networkmanager","meta_rev_id":"bf5665f4d5831ca469b9abc32a4ff157281ff131","_number":865433,"virtual_id_number":865433,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-29 13:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-11-28 16:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-29 03:18:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2022-11-28 13:38:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2022-11-29 13:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-23 14:19:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-24 22:19:58.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2022-11-28 13:38:21.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-11-28 16:56:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"908f4f37c88afd7d74492e0d6ac81731fe1350b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-11-23 14:12:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cdda5a0c3085b5308753d39891d1af0122b3fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-23 14:19:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da10723502d6457da21c26eba39d3182\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/8b973203dd8441e5b63c323c5473256b : SUCCESS in 3m 08s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4bcb36c1282e4d04b14840ebbeee7d80 : SUCCESS in 6m 00s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/b04a5601863f42f2aaddc2fb6887c885 : SUCCESS in 3m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7e795b566e8a6fee9fba92a6173e877a247adb7d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-11-24 08:18:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"772d68913364ec1875d0d79dd63ba16df43a6cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 08:25:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75ffb3f168614fce9223dec56f4c95a1\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/9aeff63f946045dc82199cc484d31c02 : SUCCESS in 3m 04s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/78cee483df6e4936b7b324b9a65fd7bf : SUCCESS in 5m 54s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/3391a3e9ed7d41e99137ef0907038d1d : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":2},{"id":"2e2399b545c50a424faef1e0ddf811acb1a8b42b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-11-24 17:11:57.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"bbd26e9f2e40ea4b5701fc28691f669af56724f7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-11-24 22:19:58.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"600b4d0e2bd98f4097fef590658c9ec006d26560","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-11-28 08:25:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4645ac996769e9c48d5640ddaf89cf2d3c2b4f65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-11-28 08:25:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"83766b3a9d53cfa3ec561f09c234085b1d3f8ee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 08:40:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eab7e064847747beb82593ec53d8dec9\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/88fb449e5ef243869eddc07afed67852 : SUCCESS in 3m 08s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6c407a28da104fbaa3f918d5070313a1 : SUCCESS in 5m 27s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/4d30e34c80614a6abe823bd88b7b1021 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"2c18ed120d746b21750ac095157b353b6df66979","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-28 13:38:21.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a8cbd029c4fa7fda619ea4ec8822edd155221644","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-11-28 16:56:52.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a2ddf9412d353237665ac4653830d4c4a87ab1a4","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-11-29 03:18:59.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ba3749643e71ba1324a07b9981d5d0633b2d821","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-29 13:12:19.000000000","message":"Patch Set 3: Workflow+1\n\nWe\u0027ll need to check the image build logs, and also make sure images built after this is deployed continue to run jobs successfully.","accounts_in_message":[],"_revision_number":3},{"id":"8e7344e0601a2a682e2663c1de569db8686619b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 13:12:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0b482b248068249911a4000249c149539b54bb25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 13:19:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ff979e561ea411982b21eee84bfe053\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1208089db3f34ce9b4a617b8772b84e4 : SUCCESS in 6m 07s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/d65b7c08910c46259b9e216a3a3c6e19 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"62c9449fa1c058e2520d32fddfa451920721401d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 13:19:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bf5665f4d5831ca469b9abc32a4ff157281ff131","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 13:44:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec7f004fa0c549d6aadb0a54b9265efc\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/3806a2f154d74c1b831e504d83847356 : SUCCESS in 10m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"368bb77ee36d3bdae42cf77e57cfaa2718f383f4","revisions":{"0ed80fc09880081148438d04ec0137e9d1e0498f":{"kind":"REWORK","_number":1,"created":"2022-11-23 14:12:54.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/33/865433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/865433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/865433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/865433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/865433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/865433/1"}}},"commit":{"parents":[{"commit":"020653d0417f34d9591a88acaacb8d9970723555","subject":"Merge \"Add post-review pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/020653d0417f34d9591a88acaacb8d9970723555"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-23 14:11:44.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-23 14:11:44.000000000","tz":60},"subject":"Ensure NetworkManager doesn\u0027t override /etc/resolv.conf","message":"Ensure NetworkManager doesn\u0027t override /etc/resolv.conf\n\nUntil now, whenever NetworkManager gets reloaded/restarted/lease\nrefresh, it would override the /etc/resolv.conf file with the\nnameservers and related it gets from the network.\n\nThis patch ensures this won\u0027t happen ever again.\n\nChange-Id: I92bc12b8f712e28962d24dd6474cfce22b81222c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0ed80fc09880081148438d04ec0137e9d1e0498f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0ed80fc09880081148438d04ec0137e9d1e0498f"}]},"branch":"refs/heads/master"},"e919a9b1171efe98aa4051700a63aa5318c63e75":{"kind":"REWORK","_number":2,"created":"2022-11-24 08:18:26.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/33/865433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/865433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/865433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/865433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/865433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/865433/2"}}},"commit":{"parents":[{"commit":"020653d0417f34d9591a88acaacb8d9970723555","subject":"Merge \"Add post-review pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/020653d0417f34d9591a88acaacb8d9970723555"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-23 14:11:44.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-24 08:18:11.000000000","tz":60},"subject":"Ensure NetworkManager doesn\u0027t override /etc/resolv.conf","message":"Ensure NetworkManager doesn\u0027t override /etc/resolv.conf\n\nUntil now, whenever NetworkManager gets reloaded/restarted/lease\nrefresh, it would override the /etc/resolv.conf file with the\nnameservers and related it gets from the network.\n\nThis patch ensures this won\u0027t happen ever again.\n\nChange-Id: I92bc12b8f712e28962d24dd6474cfce22b81222c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e919a9b1171efe98aa4051700a63aa5318c63e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e919a9b1171efe98aa4051700a63aa5318c63e75"}]},"branch":"refs/heads/master"},"368bb77ee36d3bdae42cf77e57cfaa2718f383f4":{"kind":"REWORK","_number":3,"created":"2022-11-28 08:25:31.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/33/865433/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/865433/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/865433/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/865433/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/865433/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/865433/3"}}},"commit":{"parents":[{"commit":"020653d0417f34d9591a88acaacb8d9970723555","subject":"Merge \"Add post-review pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/020653d0417f34d9591a88acaacb8d9970723555"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-23 14:11:44.000000000","tz":60},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2022-11-28 08:25:23.000000000","tz":60},"subject":"Ensure NetworkManager doesn\u0027t override /etc/resolv.conf","message":"Ensure NetworkManager doesn\u0027t override /etc/resolv.conf\n\nUntil now, whenever NetworkManager gets reloaded/restarted/lease\nrefresh, it would override the /etc/resolv.conf file with the\nnameservers and related it gets from the network.\n\nThis patch ensures this won\u0027t happen ever again.\n\nChange-Id: I92bc12b8f712e28962d24dd6474cfce22b81222c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/368bb77ee36d3bdae42cf77e57cfaa2718f383f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/368bb77ee36d3bdae42cf77e57cfaa2718f383f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
