)]}'
{"id":"openstack%2Fproject-config~865686","triplet_id":"openstack%2Fproject-config~master~I1dc5e5ed9f93ad40f2089891c63b530994b290f2","project":"openstack/project-config","branch":"master","topic":"merge-kolla-groups","attention_set":{},"removed_from_attention_set":{"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2022-11-28 16:27:46.000000000","reason":"\u003cGERRIT_ACCOUNT_6547\u003e replied on the change","reason_account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-11-29 12:50:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1dc5e5ed9f93ad40f2089891c63b530994b290f2","subject":"Use kolla.config for kolla-ansible in gerrit","status":"MERGED","created":"2022-11-25 13:35:45.000000000","updated":"2022-11-29 12:59:03.000000000","submitted":"2022-11-29 12:50:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865686-merge-kolla-groups","meta_rev_id":"1b0baf5da2ccbb6ca9065e2551645ff57daf1313","_number":865686,"virtual_id_number":865686,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-29 12:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2022-11-28 16:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-11-29 12:39:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-29 12:39:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-25 13:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-25 14:49:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-11-28 12:58:44.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2022-11-29 12:39:34.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"4df54f933a6c2ae74c7cf7ed6827ffce2bc41e9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-25 13:35:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"675a799d6e138353c05e55dfbfae8cbee1b97e08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 13:44:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2f0007f819f4e78a11598ecd04d55ca\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b98517d4bdc048bca611d1196bb03832 : SUCCESS in 2m 51s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e96597ad714d4063b9b633a9ba913b04 : SUCCESS in 6m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c06f2e18c6ae45348f646dd51f3f7ddb : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ebc4b090173c1f713b7efdcf8bc5012bd5684548","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-11-25 14:49:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"02d3a958e93174a98d67e8c7baca332770fce451","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-11-28 12:58:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"554407142a73fea395dd8fd86a1419368181fe95","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-11-28 12:59:19.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31d30747e9432a6c71aab5f76cd435fb28f9559b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-28 14:59:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"62620627a2e92400ae18dd3327af69f0e82f2099","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-11-28 15:00:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d022e6cafd13dc8d16f7ec18efd71fde0fd36cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 15:08:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1812fbdd0f88495b80eb3623e387c3fc\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/3a4a9e4044b943e5959fee6559f40c45 : SUCCESS in 3m 45s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c0e4e66ce0f5463584ab191cba15900a : SUCCESS in 6m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/047e0b01dd79428a9eb28ad87772cd03 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"74429f84877a6c016d596f285aeb0c08d58349ad","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-11-28 16:27:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"27e3cf6a308a86e4f97c816e38e4e85a65972721","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-11-29 12:39:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"abb09d5ec24047267bf74f7a372838946bfc5058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 12:39:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19264b453b426280852a9c384612fe9b6992a072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 12:50:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76749d0fb8074134b6c60af374c56c61\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e47528ae7beb4a939d64ba48d60edbf8 : SUCCESS in 6m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b393339fad584f3dac2be64769a3bb38 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"1a9f061f735ef8784930f4a1c85194b570bece59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 12:50:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1b0baf5da2ccbb6ca9065e2551645ff57daf1313","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 12:59:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d413a7fb44414e8a8c6a52b61018f173\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/c28fc066988d404e8bd879a3c6a1196a : SUCCESS in 7m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98c69f6fa9402a187ca30c29c3140e63a5050a5f","revisions":{"aba207ace7a5317ec0965b37e083c6e81485881f":{"kind":"REWORK","_number":1,"created":"2022-11-25 13:35:45.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/86/865686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/865686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/865686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/865686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/865686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/865686/1"}}},"commit":{"parents":[{"commit":"020653d0417f34d9591a88acaacb8d9970723555","subject":"Merge \"Add post-review pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/020653d0417f34d9591a88acaacb8d9970723555"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-11-25 13:30:12.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-11-25 13:30:12.000000000","tz":60},"subject":"Use kolla-core group for kolla-ansible in gerrit","message":"Use kolla-core group for kolla-ansible in gerrit\n\nThe kolla project has decided that it no longer needs to have a\ndedicated kolla-ansible-core group that is different from the kolla-core\ngroup. Membership between the two groups has already been synced, so\nthis change is essentially a no-op, but allows us to only maintain a\nsingle group going forward.\n\n[0] https://meetings.opendev.org/meetings/kolla/2022/kolla.2022-11-23-14.00.log.html\n\nChange-Id: I1dc5e5ed9f93ad40f2089891c63b530994b290f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aba207ace7a5317ec0965b37e083c6e81485881f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aba207ace7a5317ec0965b37e083c6e81485881f"}]},"branch":"refs/heads/master"},"98c69f6fa9402a187ca30c29c3140e63a5050a5f":{"kind":"REWORK","_number":2,"created":"2022-11-28 14:59:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/86/865686/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/865686/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/865686/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/865686/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/865686/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/865686/2"}}},"commit":{"parents":[{"commit":"020653d0417f34d9591a88acaacb8d9970723555","subject":"Merge \"Add post-review pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/020653d0417f34d9591a88acaacb8d9970723555"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-11-25 13:30:12.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-11-28 14:58:02.000000000","tz":60},"subject":"Use kolla.config for kolla-ansible in gerrit","message":"Use kolla.config for kolla-ansible in gerrit\n\nThe kolla project has decided that it no longer needs to have a\ndedicated kolla-ansible-core group that is different from the kolla-core\ngroup. Membership between the two groups has already been synced, so\nthis change is essentially a no-op, but allows us to only maintain a\nsingle group going forward.\n\n[0] https://meetings.opendev.org/meetings/kolla/2022/kolla.2022-11-23-14.00.log.html\n\nChange-Id: I1dc5e5ed9f93ad40f2089891c63b530994b290f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/98c69f6fa9402a187ca30c29c3140e63a5050a5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/98c69f6fa9402a187ca30c29c3140e63a5050a5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
