)]}'
{"id":"openstack%2Fproject-config~867054","triplet_id":"openstack%2Fproject-config~master~I5a6b80118afaf3b7782a1d1c131787f208583799","project":"openstack/project-config","branch":"master","topic":"fix-nodepool-api-graphs","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-12-09 13:56:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a6b80118afaf3b7782a1d1c131787f208583799","subject":"Fix nodepool API ops Grafana graphs","status":"MERGED","created":"2022-12-08 21:02:04.000000000","updated":"2022-12-09 14:04:31.000000000","submitted":"2022-12-09 13:56:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867054-fix-nodepool-api-graphs","meta_rev_id":"0b089741acd3c1566818d858102fc1c2971c200c","_number":867054,"virtual_id_number":867054,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-09 13:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-09 05:31:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2022-12-09 13:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2022-12-09 13:30:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-08 21:19:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-09 05:31:41.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2022-12-09 13:30:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"fad34c7839b6c758edc9b5cfa21a3bc6f372a524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-12-08 21:02:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff7a4039dc420d3be8b97fb7794bbeece52c2e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 21:19:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a74a1bffebcc4f39ae5d8189b31f2f56\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/d5680c485f4546fc93881edd8fa3a21e : SUCCESS in 15m 58s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/03db4864dbe145c0b8a1e35b1350ce3d : SUCCESS in 5m 46s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/3614151560d645b18527acdaf6777bff : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b92f12c8a1277f1a8eb56cfd767d0d78e0719a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-12-08 23:44:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dee7eb76215d218d161afb7efdfd401942c97194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 00:08:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b0ea1d33aff4c448005539dd245fa68\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/3856e7f93c4145ccaff717a65ddf5d60 : SUCCESS in 19m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ca67305c54f6489eac699158821413f2 : SUCCESS in 4m 57s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/f5a14773e56c4ff78a4eebe4a42372ae : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":2},{"id":"40289054a1ee47ced210482c97390511cdd793f6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-09 05:31:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"046e747889f1e433acb45d0ccd678c5974ca6c8f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-09 13:30:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bc36f1197ff6e94d322522062ca666636f208317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 13:30:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c4a0e58cca3b71307059e94aaaa38e8131f67483","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 13:56:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2052663923fe4c6d9b922021f4f10c29\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/244b51357b5f4bb588fb2ef049de1a9a : SUCCESS in 18m 28s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a21a0b7bf55244488c6ab7b168277a9b : SUCCESS in 6m 24s\n- project-config-grafana https://zuul.opendev.org/t/openstack/build/f61ca21569ae4763abddf9e485cc3621 : SUCCESS in 12m 50s","accounts_in_message":[],"_revision_number":2},{"id":"09a6f3c5fe716b09da319e5fee270bfd61364a84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 13:56:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0b089741acd3c1566818d858102fc1c2971c200c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 14:04:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fe6ec14db2c459c999ab26a1ca1fa3c\n\n- infra-prod-service-grafana https://zuul.opendev.org/t/openstack/build/99e8cd13a23c4c21b3eb9ace820d3411 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4ef6ea4d10627632bef5ee0d99e8ac53577eb0de","revisions":{"f670a971471934a8fb50c051219486958c98ea5b":{"kind":"REWORK","_number":1,"created":"2022-12-08 21:02:04.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/54/867054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/54/867054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/867054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/867054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/867054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/54/867054/1"}}},"commit":{"parents":[{"commit":"3cf341500744bc2d5e0270756f690c5288bf5c99","subject":"Merge \"Add os_skyline repo to CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3cf341500744bc2d5e0270756f690c5288bf5c99"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-12-08 21:00:23.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-12-08 21:00:23.000000000","tz":-480},"subject":"Fix nodepool API ops Grafana graphs","message":"Fix nodepool API ops Grafana graphs\n\nopenstacksdk statsd records for API operations are not response code\nspecific. This change adds a glob to the statsd record path to include\nall return code response metrics in our API graphs. I believe this\nchange may have come with the openstacksdk 0.103.0 update.\n\nWhile we are at it we stop updating the airship and inap cloud graphs\nsince those should be cleaned up and this keeps the review overhead\nsmaller.\n\nChange-Id: I5a6b80118afaf3b7782a1d1c131787f208583799\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f670a971471934a8fb50c051219486958c98ea5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f670a971471934a8fb50c051219486958c98ea5b"}]},"branch":"refs/heads/master"},"4ef6ea4d10627632bef5ee0d99e8ac53577eb0de":{"kind":"REWORK","_number":2,"created":"2022-12-08 23:44:13.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/54/867054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/54/867054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/867054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/867054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/54/867054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/54/867054/2"}}},"commit":{"parents":[{"commit":"3cf341500744bc2d5e0270756f690c5288bf5c99","subject":"Merge \"Add os_skyline repo to CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3cf341500744bc2d5e0270756f690c5288bf5c99"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-12-08 21:00:23.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-12-08 23:42:45.000000000","tz":-480},"subject":"Fix nodepool API ops Grafana graphs","message":"Fix nodepool API ops Grafana graphs\n\nopenstacksdk statsd records for API operations are not response code\nspecific. This change adds a glob to the statsd record path to include\nall return code response metrics in our API graphs. I believe this\nchange may have come with the openstacksdk 0.103.0 update. We also\nupdate paths for servers POST, server details, and flavor details as\nthey have changed.\n\nNote the network info is empty which is why we don\u0027t get graphs for\nthem, but the paths appear correct. I think this may be beacuse we don\u0027t\nneed to query network info in any of our clouds currently.\n\nWhile we are at it we stop updating the airship and inap cloud graphs\nsince those should be cleaned up and this keeps the review overhead\nsmaller.\n\nChange-Id: I5a6b80118afaf3b7782a1d1c131787f208583799\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ef6ea4d10627632bef5ee0d99e8ac53577eb0de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ef6ea4d10627632bef5ee0d99e8ac53577eb0de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
