)]}'
{"id":"openstack%2Fproject-config~867268","triplet_id":"openstack%2Fproject-config~master~Ia44a2ef83cdc57d0bb74c72ea5c95078b53ccaec","project":"openstack/project-config","branch":"master","topic":"toxv4","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-12-12 21:06:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia44a2ef83cdc57d0bb74c72ea5c95078b53ccaec","subject":"Update tox.ini for tox v4","status":"MERGED","created":"2022-12-12 17:27:59.000000000","updated":"2022-12-12 21:08:05.000000000","submitted":"2022-12-12 21:06:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867268-toxv4","meta_rev_id":"38874adcd6ff9a9595962e709503790b9656480a","_number":867268,"virtual_id_number":867268,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 21:06:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-12 20:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2022-12-12 20:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2022-12-12 20:56:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-12 17:36:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-12 20:53:21.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2022-12-12 20:56:55.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"66bc9e1f5ca014d83b9aaa53660e3fe89b74d95d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-12-12 17:27:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"122a8200f5576b10c46cb51e2f17db40338a0e2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 17:36:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/451c3d455f4a4354bb1f4afab6ae8289\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/3b8fb11b4a5a403abc7e0381ec2cb42e : SUCCESS in 2m 46s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/bf11bea3a8eb4ea5a352e3ce5ca647e6 : SUCCESS in 3m 29s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d7aa3b8d57294dc8b4a483c5e359229e : SUCCESS in 6m 09s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/ad84d94fdff447d69ca1bbb7521052bb : SUCCESS in 4m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d0db9e3d016e442a8dd6071238a2cd3e : SUCCESS in 3m 46s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/ceba284b55f64859b9f69cc72993fb4b : SUCCESS in 4m 30s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/dcecca7e1bd84171b5c107a67223c890 : SUCCESS in 5m 21s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/d9eaaba54079493f82ae4d38e0076a0c : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"1f287849469302c176ad37d66f84fc340b3c55b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-12-12 18:29:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"13cc59dbaf3b6275eb88515d6e8051d24b794871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 18:44:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4904e8db39dd4955b9f876c4d7d99ddd\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/77b7d949541142218f24ed72f2ca7a1d : SUCCESS in 2m 51s\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/f0527afadb2648f4a1d35e16046412f8 : SUCCESS in 3m 03s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4517146de5f143728e2d16fb47245b6b : SUCCESS in 7m 01s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/3450f7c9ecc24ff88dc06f8e103b2329 : SUCCESS in 3m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/30c4a7c7a6b74f479366649cc21505bf : SUCCESS in 3m 30s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/fb569d94b4434089b415ddda9bde63b8 : SUCCESS in 3m 39s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/e4750d29445240958d8fe6aec5e45ad6 : SUCCESS in 5m 29s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/7ae995cbd555493ab91ba3367f9eb7bc : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":2},{"id":"d9dfab84ab82aaffe4da67d7c9143486ee563b08","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-12 20:53:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a63b6fbd5a5b7c5494f443db8101d2774d9369c8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-12 20:56:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99ef6d18469d406b36d357bc5a59d3f7ea9ce27c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 20:57:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"48c9c2a6ab269f0fff609cd55876e0243bc9a4fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 21:06:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88b3216d7aa24d62b0de2ce626e8257b\n\n- build-openstack-specs-site https://zuul.opendev.org/t/openstack/build/ba69d15788eb4072b8f6fc693316a713 : SUCCESS in 3m 46s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f9ca494df54f4804971807ce396d899b : SUCCESS in 6m 18s\n- project-config-dib https://zuul.opendev.org/t/openstack/build/b03f2a8643844fd39bd0ba35c1327bb6 : SUCCESS in 3m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/34a856ccbf59425e8841ede19ed0f8ab : SUCCESS in 3m 34s\n- project-config-infra-docs-index https://zuul.opendev.org/t/openstack/build/01b8ec66518a4efeaa1f7a61df74f7a4 : SUCCESS in 3m 45s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/3c7f4688ecbf496f9ad881dc5453861c : SUCCESS in 5m 20s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/a3361c84bf004557a15259c86e0a6a9e : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"56b34d07da34b8125d1e0356b3643b881e09bde1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 21:06:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"38874adcd6ff9a9595962e709503790b9656480a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 21:08:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a52f88c9b8a1458ba485a5033d6a8a31\n\n- promote-infra-index https://zuul.opendev.org/t/openstack/build/72c26215988342048cc8ba2fe6271ca3 : SUCCESS in 46s\n- promote-openstack-specs-site https://zuul.opendev.org/t/openstack/build/c8627845dc96496681737ae008fdd455 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7af758c33de1ac60774c2008cb550a091a8bd75","revisions":{"055bff5a6a4c43d01d8e9a8e57e3f99512ae928c":{"kind":"REWORK","_number":1,"created":"2022-12-12 17:27:59.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/68/867268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/867268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/867268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/867268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/867268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/867268/1"}}},"commit":{"parents":[{"commit":"e2a62415cea7ff617cc85e63deb663c07a29b684","subject":"Merge \"Fix nodepool API ops Grafana graphs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2a62415cea7ff617cc85e63deb663c07a29b684"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-12-12 17:26:48.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-12-12 17:27:52.000000000","tz":-480},"subject":"Update tox.ini for tox v4","message":"Update tox.ini for tox v4\n\nThis convers whitelist_externals to allowlist_externals to make tox v4\nhappy. Note that this may not be a complete fix for tox v4 as\nproject-config isn\u0027t tested speculatively which makes it harder to find\nall the cases.\n\nChange-Id: Ia44a2ef83cdc57d0bb74c72ea5c95078b53ccaec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/055bff5a6a4c43d01d8e9a8e57e3f99512ae928c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/055bff5a6a4c43d01d8e9a8e57e3f99512ae928c"}]},"branch":"refs/heads/master"},"f7af758c33de1ac60774c2008cb550a091a8bd75":{"kind":"REWORK","_number":2,"created":"2022-12-12 18:29:27.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/68/867268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/867268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/867268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/867268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/867268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/867268/2"}}},"commit":{"parents":[{"commit":"e2a62415cea7ff617cc85e63deb663c07a29b684","subject":"Merge \"Fix nodepool API ops Grafana graphs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2a62415cea7ff617cc85e63deb663c07a29b684"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-12-12 17:26:48.000000000","tz":-480},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2022-12-12 18:29:20.000000000","tz":-480},"subject":"Update tox.ini for tox v4","message":"Update tox.ini for tox v4\n\nThis convers whitelist_externals to allowlist_externals to make tox v4\nhappy. Note that this may not be a complete fix for tox v4 as\nproject-config isn\u0027t tested speculatively which makes it harder to find\nall the cases.\n\nChange-Id: Ia44a2ef83cdc57d0bb74c72ea5c95078b53ccaec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7af758c33de1ac60774c2008cb550a091a8bd75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f7af758c33de1ac60774c2008cb550a091a8bd75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
