)]}'
{"id":"openstack%2Fproject-config~867651","triplet_id":"openstack%2Fproject-config~master~I7cdac3807a8f5097bf12fd76b38721f998a5be26","project":"openstack/project-config","branch":"master","topic":"cloudkitty-api-ref","attention_set":{},"removed_from_attention_set":{"25277":{"account":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"last_update":"2023-01-10 16:16:03.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2023-01-10 16:08:03.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I7cdac3807a8f5097bf12fd76b38721f998a5be26","subject":"Add the \"api-ref-jobs\" template to CloudKitty","status":"MERGED","created":"2022-12-14 12:31:38.000000000","updated":"2023-01-10 16:16:03.000000000","submitted":"2023-01-10 16:16:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867651-cloudkitty-api-ref","meta_rev_id":"8a7860231e1f3222daf44c6b80e993d1d7682df3","_number":867651,"virtual_id_number":867651,"owner":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-10 16:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2023-01-04 09:54:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2023-01-04 11:39:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-14 17:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-10 16:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-14 12:39:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-14 12:54:32.000000000","updated_by":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"reviewer":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"state":"REVIEWER"},{"updated":"2022-12-14 13:01:57.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2022-12-14 16:38:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"7747d8a4952ff5856b99b00e96f45e73ba6fa89c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2022-12-14 12:31:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25911f76f757a0facd11a136631fa3d34c1df998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 12:39:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e00c2ba6a1e41b1a5f402990cb2fdcf\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ee80ec35294a43ecb781f7d5f1da116e : SUCCESS in 2m 49s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1de2031f26c54217a6ba5e7489a1d45c : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"bce19f9ac7c3c99e5096b7c2201519062d0b4408","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-12-14 12:54:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"be500105b3101748b945bbc518897e03884f4623","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2022-12-14 13:01:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eee44e4a5fab43f92264b454c5ce9afe271be995","author":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"date":"2022-12-14 13:25:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ba31b322ec6e6df166349f6ce6585658d340c4c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-14 16:38:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a2472d380f9af3b697ba9dad28558a8fcd1d1aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 16:38:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d794bc0ea2f60b218a81016eab5ba576661a7c32","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-14 16:39:25.000000000","message":"Patch Set 1: -Workflow\n\nOops, unapproving until the undeclared dependency has merged.","accounts_in_message":[],"_revision_number":1},{"id":"d04548f194c0b9bb5bb49c59477143f4c6a74c39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 16:39:37.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"3421bcd758d617d3da82fcedee696e5001815a64","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-14 16:40:05.000000000","message":"Patch Set 1:\n\nIn future, you can set a Depends-On footer within the commit message to make sure they merge in the correct order.","accounts_in_message":[],"_revision_number":1},{"id":"9b7d9596c459a2eb713bfa2cc1e94beb975e36b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-14 16:41:29.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3721380f75cbcf03e25c26a4096d7e32e6062bcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-14 16:48:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2a53a2ef8040ccb55637cd15d523db\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/b0fed9a839b948a5b7c3ad498de1087d : SUCCESS in 3m 00s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/536952cfe6a44783b1c6cf3634d27a26 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"1f47eecbcbb350bf4290654539724f0ddabc85e5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-14 17:10:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"adc3e1a453eee2d41de5257c845f4fcb6db77273","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-01-04 09:54:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1c1b1d1cced284ba9ed9521ec10f6d7209e5a5b9","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-01-04 11:39:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ee294326b85cc060497734cd911281b3b73fa2f4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-10 16:08:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6f5248b5a65734e551f58a0d4eb76bd7378f6210","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 16:08:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e907f72fd9f81d42ee899fcebc343eb81d13f095","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 16:16:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f70e12228d2490a985218b1570013ab\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/df936e6051c245798125f59e38cfb5ed : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"8a7860231e1f3222daf44c6b80e993d1d7682df3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 16:16:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cb012cdb2bb3ffa1990b36ecf081f9af90ff9220","revisions":{"9f25314f77367d25140c96e8f9c33744b0fea52d":{"kind":"REWORK","_number":1,"created":"2022-12-14 12:31:38.000000000","uploader":{"_account_id":25277,"name":"Mariusz Karpiarz","email":"mariusz.karpiarz@codethink.co.uk","username":"mkarpiarz"},"ref":"refs/changes/51/867651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/867651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/867651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/867651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/867651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/867651/1"}}},"commit":{"parents":[{"commit":"f540b4cac0bb8a74a53a0445158aac4768fd489c","subject":"Merge \"Ensure NetworkManager doesn\u0027t override /etc/resolv.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f540b4cac0bb8a74a53a0445158aac4768fd489c"}]}],"author":{"name":"Mariusz Karpiarz","email":"m.karpiarz@eschercloud.ai","date":"2022-12-14 12:29:38.000000000","tz":0},"committer":{"name":"Mariusz Karpiarz","email":"m.karpiarz@eschercloud.ai","date":"2022-12-14 12:29:38.000000000","tz":0},"subject":"Add the \"api-ref-jobs\" template to CloudKitty","message":"Add the \"api-ref-jobs\" template to CloudKitty\n\nChange-Id: I7cdac3807a8f5097bf12fd76b38721f998a5be26\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9f25314f77367d25140c96e8f9c33744b0fea52d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9f25314f77367d25140c96e8f9c33744b0fea52d"}]},"branch":"refs/heads/master"},"cb012cdb2bb3ffa1990b36ecf081f9af90ff9220":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-14 16:41:29.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/51/867651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/867651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/867651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/867651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/867651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/867651/2"}}},"commit":{"parents":[{"commit":"f540b4cac0bb8a74a53a0445158aac4768fd489c","subject":"Merge \"Ensure NetworkManager doesn\u0027t override /etc/resolv.conf\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f540b4cac0bb8a74a53a0445158aac4768fd489c"}]}],"author":{"name":"Mariusz Karpiarz","email":"m.karpiarz@eschercloud.ai","date":"2022-12-14 12:29:38.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-12-14 16:41:29.000000000","tz":0},"subject":"Add the \"api-ref-jobs\" template to CloudKitty","message":"Add the \"api-ref-jobs\" template to CloudKitty\n\nChange-Id: I7cdac3807a8f5097bf12fd76b38721f998a5be26\nDepends-On: https://review.opendev.org/867628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb012cdb2bb3ffa1990b36ecf081f9af90ff9220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb012cdb2bb3ffa1990b36ecf081f9af90ff9220"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
