)]}'
{"id":"openstack%2Fproject-config~867801","triplet_id":"openstack%2Fproject-config~master~Idbf614be4b88db56258ed3b0e7725b581196a6fb","project":"openstack/project-config","branch":"master","topic":"bring-ptlplusone-back","attention_set":{},"removed_from_attention_set":{"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2022-12-15 15:34:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbf614be4b88db56258ed3b0e7725b581196a6fb","subject":"Bring back the PTL+1 column in release dashboard","status":"MERGED","created":"2022-12-15 13:00:49.000000000","updated":"2022-12-15 15:42:30.000000000","submitted":"2022-12-15 15:34:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867801-bring-ptlplusone-back","meta_rev_id":"fbb2a85f694274fa1b9292d3c4c776622b943475","_number":867801,"virtual_id_number":867801,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-15 15:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2022-12-15 15:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-15 13:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2022-12-15 15:25:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 13:04:44.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-12-15 13:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-15 15:25:18.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"1aae667a3da90ed7554f580b7c931eb4456b63c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-12-15 13:00:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b7a1b96479b92f49c85d87dcbb56effc1cfcd4c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-12-15 13:04:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acf7b311a9851703a59e50d90654dde00677251e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 13:11:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb637301b0784a58a16e08043b5c2d6f\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/2cf3eac39add43979fdbd6b5c243f1dd : SUCCESS in 2m 33s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/578c36114e5148bbb21c12c4d2f523f2 : SUCCESS in 5m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/77a50d9b98c74d8391bc58f48fe35fc7 : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"ca38894c2d6a2f7650c20f6e82c6cc3bb525cca0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-12-15 15:25:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f28383bf9f6cce74348788efc4ff46e932b19db0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 15:25:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c608448773a785dfd5759b074a77ce6137cee374","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 15:34:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8941ffb442a147fdb54bb8a400547513\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1d349f81a993497d9fbab8eeab21a623 : SUCCESS in 5m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5cb4515f5e854c928f066ced4cf63141 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fecaa8f473f6941bb7f7690c12fe56275e410071","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 15:34:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fbb2a85f694274fa1b9292d3c4c776622b943475","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 15:42:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c32c9ab019e94340893e25dc5d292bd9\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/12a488ef1d9e46c582b76f3ae64c64b3 : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de372833c5998f4b3a9c6d804e647e7da212c631","revisions":{"de372833c5998f4b3a9c6d804e647e7da212c631":{"kind":"REWORK","_number":1,"created":"2022-12-15 13:00:49.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/01/867801/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/867801/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/867801/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/867801/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/867801/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/867801/1"}}},"commit":{"parents":[{"commit":"91a9c8be67943f6a0b5c1b1f6d7c19c091787f99","subject":"Merge \"Update post-review zuul pipeline definiton\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/91a9c8be67943f6a0b5c1b1f6d7c19c091787f99"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2022-12-15 12:56:17.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2022-12-15 12:56:17.000000000","tz":60},"subject":"Bring back the PTL+1 column in release dashboard","message":"Bring back the PTL+1 column in release dashboard\n\nFollowing the Gerrit upgrade, the release team lost the \"PTL+1\" column\nin the release dashboard, which made it convenient dte screen\nready-to-process changes.\n\nNew gerrit only displays \"required\" labels. This change makes PTL+1\nblocking in case the lowest negative value is selected (AnyWithBlock),\nwhile defining a negative value that nobody can actually set.\n\nChange-Id: Idbf614be4b88db56258ed3b0e7725b581196a6fb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de372833c5998f4b3a9c6d804e647e7da212c631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de372833c5998f4b3a9c6d804e647e7da212c631"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
