)]}'
{"id":"openstack%2Fproject-config~868441","triplet_id":"openstack%2Fproject-config~master~I5c7d9bfc7fc3ed47196d00dfc68350b74200f6b4","project":"openstack/project-config","branch":"master","topic":"linaro-2022","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-01-04 13:38:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5c7d9bfc7fc3ed47196d00dfc68350b74200f6b4","subject":"nodepool: remove ubuntu-bionic-arm64-large label","status":"MERGED","created":"2022-12-22 22:12:47.000000000","updated":"2023-01-04 13:47:12.000000000","submitted":"2023-01-04 13:38:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868441-linaro-2022","meta_rev_id":"7a2ef8e433558cbc3e4577189a4a8ebeecdade09","_number":868441,"virtual_id_number":868441,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-04 13:38:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2023-01-04 08:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-04 13:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-04 13:31:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-22 22:26:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 08:55:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2023-01-04 13:31:14.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"422d134aee00f6776b19855243742341bf161a8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:12:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c91a171144bb4a229de78094292f691c590ed31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 22:26:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a88e55dff384a639499ecf63803f7f3\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/cfb59ad3689c4933bc53710cd65756a4 : SUCCESS in 2m 44s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ec4a8edbe9d44d88bf1a78508290b505 : SUCCESS in 7m 12s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/3a1350374ad94e0b96d14de81b58cfd3 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e6eeaebedd7ed0f8c6b5377579b11ecb8036c255","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:56:57.000000000","message":"Topic linaro-nodepool removed","accounts_in_message":[],"_revision_number":1},{"id":"c005b27c8a630cb761c9c1885498b9803468c5a0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:57:01.000000000","message":"Topic set to linaro-2022","accounts_in_message":[],"_revision_number":1},{"id":"aafa797ba880b225bb254ee4f0ea87c37d6707dd","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2023-01-04 08:55:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d45c5daf6c4fbf014625b7eb0963699cf4b062e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-04 13:31:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f85bde3dddc332ff9efccf9438c9c500d558e01c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 13:31:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84e5b975c1b90ef73719f9cba94693200722f589","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 13:38:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ff640f7548840a29e464a8647b5d675\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2312076c15124491a2a3ffbcbda4ab5a : SUCCESS in 6m 44s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/05e7ea7734844b5b8523bda1bfd9c179 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a2c0fb70df18aa1a960e9aee4a24502749dbdc53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 13:38:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a2ef8e433558cbc3e4577189a4a8ebeecdade09","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-04 13:47:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb49ec161dad47ab87c88ee86709c3a1\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/74e15ada1c08401284bae00ca9ff4cd3 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcad94f4e1bedcddd669a029fe07382bff988815","revisions":{"fcad94f4e1bedcddd669a029fe07382bff988815":{"kind":"REWORK","_number":1,"created":"2022-12-22 22:12:47.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/41/868441/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/868441/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/868441/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/868441/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/868441/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/868441/1"}}},"commit":{"parents":[{"commit":"4bc39369453896d4fa52ecd80244ae2134657497","subject":"Update Grafana\u0027s dasboard for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4bc39369453896d4fa52ecd80244ae2134657497"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-12-22 21:41:45.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-12-22 21:43:45.000000000","tz":660},"subject":"nodepool: remove ubuntu-bionic-arm64-large label","message":"nodepool: remove ubuntu-bionic-arm64-large label\n\nThis was only used for gate testing and nothing committed uses this\nnode type.  None of the other more recent distros have a\ncorreseponding -large type.  Remove this to cleanup the node types and\nreduce required flavors.\n\nChange-Id: I5c7d9bfc7fc3ed47196d00dfc68350b74200f6b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcad94f4e1bedcddd669a029fe07382bff988815"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fcad94f4e1bedcddd669a029fe07382bff988815"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[]}
