)]}'
{"id":"openstack%2Fproject-config~868442","triplet_id":"openstack%2Fproject-config~master~I8c8439ca2ae8dd868bce9d2b015f6b8428b16a84","project":"openstack/project-config","branch":"master","topic":"linaro-2022","attention_set":{},"removed_from_attention_set":{"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-01-12 14:25:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c8439ca2ae8dd868bce9d2b015f6b8428b16a84","subject":"Add new linaro cloud","status":"MERGED","created":"2022-12-22 22:12:47.000000000","updated":"2023-01-12 14:51:18.000000000","submitted":"2023-01-12 14:25:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868442-linaro-2022","meta_rev_id":"390b152543fd2f5c18107b44111c6ddc751bf749","_number":868442,"virtual_id_number":868442,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 14:25:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2023-01-10 19:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-04 13:33:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2023-01-10 19:45:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-22 22:26:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 13:33:28.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-01-10 19:43:08.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"1e6a8b37948d09417f3bfc3a027383ecb576b5ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:12:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cf5f868867ecebfee4805123579f2e8ddfe54c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 22:26:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a62339c02c44d2492bec378a6589663\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/ffc137f129164bcc9851674cc9e4344a : SUCCESS in 2m 47s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/057776560016425e9bcb4701ac3e546c : SUCCESS in 7m 11s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/deff5c8e247f4535af407ff23dbd8891 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"30c99f14119db74212dbc625421bcfa2b1ddcccc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:56:49.000000000","message":"Topic linaro-nodepool removed","accounts_in_message":[],"_revision_number":1},{"id":"22de095b0d988e3384e235da9ae7ba4564b45dc8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:56:54.000000000","message":"Topic set to linaro-2022","accounts_in_message":[],"_revision_number":1},{"id":"33fb05a51e80712d29e3855da3b53ba2668cd91d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-04 13:33:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"075eab4d0e92d679381be93840d6e6f4037640b7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-10 19:43:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26b6f271b08ccec1aa9ac5af0f6309d47cee4311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 19:43:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25b021deca3342073aa53977303e63851b1ddbf3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-10 19:44:51.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"858630699a5dc41d6cbe768032282f0bc7a251ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 19:45:04.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":1},{"id":"18cc448598a63d4522d0dbaa7a69941381f61f86","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-10 19:45:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8b3d56f54f12e1e809aab367357022e0cc41262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-10 19:51:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45d02fe04d344f31a70667c1ba2eefd2\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/c4945ec30a9340f1a56728211024135b : SUCCESS in 2m 45s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8476aefe0c1249a1843f475899b144ed : FAILURE in 5m 13s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/f1fcf59f5ed24301a2c207d092964ac1 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"424163729a486b1ab781070d2ac6306abea3625b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-12 13:58:16.000000000","message":"Patch Set 1:\n\nrecheck now that 92d9cd7d8d994c42d880f4b735fc8119ceec8888 has the linter fix","accounts_in_message":[],"_revision_number":1},{"id":"57edc27c3d422d9d7b4d8088a82c8dfa6de50921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:10:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02cc09ed96c54f6fafe2793f29921f1c\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/5ea0c7cc27fa4aa48c6894229c581b07 : SUCCESS in 3m 07s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e7c6db9bd2bf493d97cd07f38737d309 : SUCCESS in 5m 27s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/c57dc2ac2b904b379f0cb0834b97f695 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fd736c3decd34361086b052a7d7129ee42eee1e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:11:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e1c9293aa1bad7dfcd3bf1403038d0655791a50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:25:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af0657b67a474c6f8087ca94dea74dcc\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e2c950c9bcc545969cdb763c88a5e6f8 : SUCCESS in 5m 53s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2ea357a0bbbf4d1f9a14b98f3e1d7a26 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9f2960bb64e1c73652a1cc6f22ab321a4ca965e9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:25:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"390b152543fd2f5c18107b44111c6ddc751bf749","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:51:18.000000000","message":"Patch Set 1:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28e1a1c150314b5d9b33b9d3cbaee090\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/104760a4f46c4952acd781060a246935 : FAILURE in 8m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccde55cb469205d953b9ca288a2e87918d4c4858","revisions":{"ccde55cb469205d953b9ca288a2e87918d4c4858":{"kind":"REWORK","_number":1,"created":"2022-12-22 22:12:47.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/42/868442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/42/868442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/868442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/868442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/868442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/42/868442/1"}}},"commit":{"parents":[{"commit":"5a586b3751c1b73c737573aa6a75563d2291dee2","subject":"ARM64 nodes : rename larger types to be more descriptive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a586b3751c1b73c737573aa6a75563d2291dee2"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-12-22 21:40:38.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-12-22 21:56:43.000000000","tz":660},"subject":"Add new linaro cloud","message":"Add new linaro cloud\n\nAdd nodes for the new linaro cloud\n\nDepends-On: https://review.opendev.org/c/opendev/system-config/+/868433\nChange-Id: I8c8439ca2ae8dd868bce9d2b015f6b8428b16a84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccde55cb469205d953b9ca288a2e87918d4c4858"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccde55cb469205d953b9ca288a2e87918d4c4858"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[]}
