)]}'
{"id":"openstack%2Fproject-config~868443","triplet_id":"openstack%2Fproject-config~master~I4225a5d58e2988a3e2da4799c95df9f9e39fe1e2","project":"openstack/project-config","branch":"master","topic":"linaro-nodepool","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-01-16 02:25:32.000000000","reason":"Change was submitted"},"7118":{"account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"last_update":"2023-01-16 00:51:16.000000000","reason":"\u003cGERRIT_ACCOUNT_7118\u003e replied on the change","reason_account":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}},"hashtags":[],"change_id":"I4225a5d58e2988a3e2da4799c95df9f9e39fe1e2","subject":"nl03 : reorganise labels","status":"MERGED","created":"2022-12-22 22:12:47.000000000","updated":"2023-01-16 03:13:19.000000000","submitted":"2023-01-16 02:25:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868443-linaro-nodepool","meta_rev_id":"4a2f0fe6742f362da361e38b8f5995616d5004d7","_number":868443,"virtual_id_number":868443,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-16 02:25:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-01-16 02:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2023-01-16 02:28:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-01-16 02:07:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-22 22:20:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-04 13:37:50.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2023-01-10 19:43:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"aaae8002478d07b28fb4b69e2f0c145a758a3d89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:12:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47d8b5ffdfec912e87ae638d0ae3330fa3ebda94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 22:20:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8762b570ac944b56849afbd5a2fc78c7\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/e23c535929b14790b0df31eddfaa256d : SUCCESS in 2m 50s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/393b87f21125417e902ae44079a60820 : SUCCESS in 4m 51s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/39bc1f49e5274726b6a7c9fc020c2317 : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"9402a83c4dd7df2a880b33d59239a43b84c637ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:46:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d268fe0d0a2e5023e42a323e57c6d715b109bc03","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:56:24.000000000","message":"Topic linaro-nodepool removed","accounts_in_message":[],"_revision_number":2},{"id":"1f8df6dab068985fc996d21a411083380514cc23","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2022-12-22 22:56:45.000000000","message":"Topic set to linaro-2022","accounts_in_message":[],"_revision_number":2},{"id":"29847794587b23c5f7cdde6093d50ad54fa58e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-22 22:58:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17397bbceead41d3a4682129056aacd5\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/7e8d326795224b4cb99fec8ef7450d91 : SUCCESS in 2m 46s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f1c6e330fa904fb48600e0ca9aa01c99 : SUCCESS in 5m 32s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/001da95ec5e14f2195bd8bf7098ec667 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"6ac57d0d93abb20f4e667c7323cc2f201e9c55c9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-04 13:37:50.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, this ought to make these entries much easier to manage if we can keep them grouped this way.","accounts_in_message":[],"_revision_number":2},{"id":"5ded33b50dbf8f731fbb7a86ea9fafad73af7c22","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-01-10 19:43:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4bdf2492c1b5cd0fe6942f2bdda58cf9f6718a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-16 00:51:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f1ce82a6622c22ed9c957db4ecd7e21cbc893eee","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-16 00:51:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a296bae747888398d31e8b4a5f365bead8378e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 00:57:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1337bc1d02a743c189538d5d5b7a89f8\n\n- opendev-buildset-registry https://zuul.opendev.org/t/openstack/build/727f7fca2b58402f83a2a7010c6b1270 : SUCCESS in 2m 57s\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/26633c8de34648ffb08a2dce0339164d : SUCCESS in 5m 05s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/47a412f2fd5c4695a604a5fb28a0de6c : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"605c66ebc651ef1fcc04726147cfcd4d5f0408ab","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2023-01-16 02:07:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e9599e3250fafbc4ae07f11c84cadc99bfa0416f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 02:12:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f8a675f10c1fa46e91398305a74f810b0a66710b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 02:25:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06f5cf95db6441c8b02f47743669d605\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d9d356fe10f544b19b0f3689d9d7da01 : SUCCESS in 6m 24s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/ba15fb50778b4e5d8eb406d51d1d741b : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":3},{"id":"fe2ac53233c29376d59c21a0849721c938a284c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 02:25:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"26db62ef48e14c7e3218b48ac738596fe6f2bbe6","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-01-16 02:28:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4a2f0fe6742f362da361e38b8f5995616d5004d7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 03:13:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/829e1cf3a9974f81b68c40e1770db3c7\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/2cc1e293af33418aa27b0fa5aa89f4d9 : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1a855f3e9cd54988b1fb2199c85a98fb718fcffd","revisions":{"6f44d51a4084146fcdcf02b97bf18059b2e0829b":{"kind":"REWORK","_number":1,"created":"2022-12-22 22:12:47.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/43/868443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/868443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/868443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/868443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/868443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/868443/1"}}},"commit":{"parents":[{"commit":"ccde55cb469205d953b9ca288a2e87918d4c4858","subject":"Add new linaro cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccde55cb469205d953b9ca288a2e87918d4c4858"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-12-22 22:10:43.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-12-22 22:10:43.000000000","tz":660},"subject":"nl03 : reorganise labels","message":"nl03 : reorganise labels\n\nReorganise these labels into x86_64, arm64 and vexxhost specific.  I\nthink grouping by arch is a more logical grouping of what is going on\nfor the usual operations on this file.\n\nChange-Id: I4225a5d58e2988a3e2da4799c95df9f9e39fe1e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f44d51a4084146fcdcf02b97bf18059b2e0829b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f44d51a4084146fcdcf02b97bf18059b2e0829b"}]},"branch":"refs/heads/master"},"b5674f0d64d1dc100c90951ca4c63677ed32dd31":{"kind":"REWORK","_number":2,"created":"2022-12-22 22:46:38.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/43/868443/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/868443/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/868443/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/868443/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/868443/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/868443/2"}}},"commit":{"parents":[{"commit":"ccde55cb469205d953b9ca288a2e87918d4c4858","subject":"Add new linaro cloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccde55cb469205d953b9ca288a2e87918d4c4858"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-12-22 22:10:43.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-12-22 22:46:24.000000000","tz":660},"subject":"nl03 : reorganise labels","message":"nl03 : reorganise labels\n\nReorganise these labels into x86_64, arm64 and vexxhost specific.  I\nthink grouping by arch is a more logical grouping of what is going on\nfor the usual operations on this file.\n\nChange-Id: I4225a5d58e2988a3e2da4799c95df9f9e39fe1e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b5674f0d64d1dc100c90951ca4c63677ed32dd31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b5674f0d64d1dc100c90951ca4c63677ed32dd31"}]},"branch":"refs/heads/master"},"1a855f3e9cd54988b1fb2199c85a98fb718fcffd":{"kind":"REWORK","_number":3,"created":"2023-01-16 00:51:02.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/43/868443/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/43/868443/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/868443/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/868443/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/43/868443/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/43/868443/3"}}},"commit":{"parents":[{"commit":"505b7c1479456df1add18ee9cabb57e8a2146a14","subject":"Merge \"Add new linaro cloud\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/505b7c1479456df1add18ee9cabb57e8a2146a14"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2022-12-22 22:10:43.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2023-01-16 00:50:50.000000000","tz":660},"subject":"nl03 : reorganise labels","message":"nl03 : reorganise labels\n\nReorganise these labels into x86_64, arm64 and vexxhost specific.  I\nthink grouping by arch is a more logical grouping of what is going on\nfor the usual operations on this file.\n\nChange-Id: I4225a5d58e2988a3e2da4799c95df9f9e39fe1e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a855f3e9cd54988b1fb2199c85a98fb718fcffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1a855f3e9cd54988b1fb2199c85a98fb718fcffd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[]}
